)]}'
{"id":"openstack%2Fkayobe~687499","triplet_id":"openstack%2Fkayobe~master~I3a2c9ae392745e001c5bdd01212c996e54c44f0c","project":"openstack/kayobe","branch":"master","topic":"sphinx-requirements","hashtags":[],"change_id":"I3a2c9ae392745e001c5bdd01212c996e54c44f0c","subject":"Remove sphinx from test-requirements.txt","status":"MERGED","created":"2019-10-09 08:55:32.000000000","updated":"2019-12-06 18:04:20.000000000","submitted":"2019-12-06 18:04:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"687499-1575655460752-0f63802f","meta_rev_id":"a3bdf827896320abb5f1d7acc89e961e561064d2","_number":687499,"virtual_id_number":687499,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 18:04:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-06 15:18:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-10-25 11:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-21 09:11:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-12-06 15:18:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 09:11:57.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-25 11:17:53.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-06 15:18:40.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-06 18:04:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07d1df7f9c80c2a10cf81f3e1400a7c5e6d36c0a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-09 08:55:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"da8ab3b730861f237be4e77d4411b999b6155d8f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-09 09:58:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5fc97986f99d9c3b46872bfe6cb8bdb52bc42dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-09 10:58:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/930db28b2505455eb63b8f211b574645 : SUCCESS in 3m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/241f273a92234dc890016effe3f9b0b7 : POST_FAILURE in 16m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edc5b2a8c6234607a7906c699136fd98 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93357c6ea4e34982be55731a10d504e9 : FAILURE in 3m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70c4709036344faebff6a3ba68b9ae4a : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4f20a0c82bd44f88db54a50a999bb87 : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb445c0f273546c5b73f6118000c5a78 : SUCCESS in 3m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c776ce803984d479fe62df28eb13058 : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f60f0218497a4c0787f9dbfbe73569f0 : TIMED_OUT in 30m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b9eb3cdfc13a4c469dcc959d20a41a7c : SUCCESS in 52m 14s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/98470e7da3a542b09e035afde8a831f4 : SUCCESS in 1h 10m 10s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/37af6bb933cf4c87bf3bad1c6e6c5c16 : SUCCESS in 1h 58m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0a26038ab8be405bbd3e9dc57cd403d7 : SUCCESS in 20m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a3a40421763e4bfb9fba91b752db12e9 : SUCCESS in 38m 38s","accounts_in_message":[],"_revision_number":1},{"id":"08dfd8f259f706be48884f0f370dfbd0f63c6491","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-09 13:56:40.000000000","message":"Patch Set 1: Workflow-1\n\ntox pep8 job runs sphinx8 which needs these requirements.","accounts_in_message":[],"_revision_number":1},{"id":"02be5e231743b776c52446d95b5f9f220c78c6b4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-21 08:55:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"9d0c67e1f54d0ea0336a3001e97f79a11f04e60d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-21 09:04:03.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"7596b7cd4d59f10acba7cf264bd288db053530f9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-21 09:10:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a241bfa3d2e757f788e79394d01531f01812d9b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 09:11:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d8993a3ca64dd359627e668d4590e0dbd8fc4ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 11:22:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf24377ea5d5481f8973aef2f7132a3e : SUCCESS in 5m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e6810facb50a49cdaeed034c86adc772 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/336edfb1734c411bafb7fea9b18e3cb6 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5214a92f2ac4246a78a6a5278959ee2 : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c5d03a62399493f9197762596c178da : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f364adaa0dba47cf8601ad8084c04853 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01970151cd014d719521a5506572d543 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/80188ebb09fb4689bbd493959b248fe2 : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/69960950d8ba4006a7d3c8b94f44ba03 : SUCCESS in 7m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e895994b9d82483d9988458054f5ff35 : SUCCESS in 38m 42s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d9733058cdfd451b8cc00f8e50705bec : FAILURE in 16m 35s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/78cbda5631ee4deabdc0c8266cf9dd8f : FAILURE in 2h 09m 29s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c5515b80341341d182dd363c0825ab5c : SUCCESS in 17m 11s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bea40a5203d74cd9938d985711b6e214 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":2},{"id":"400625e00b687cab3dddc240993a0d77da5d355e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 11:17:53.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"af335560e798b359919c43a70419f5bd45156116","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:20:16.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b099730125541e7ad132459b4822b49 : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1d0603b5f6fe4caa83b595cf0b337a18 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/324fcd079b1f4dbaab26d83fd75d3c5c : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd693395a85943eebbf65820413b20bf : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/439425dfb1a74890a07fe25c2c448a7b : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e188aebb4334afcb3e2c2ff4b4884aa : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02acec1f7aab4bbf8aac992186ecb208 : SUCCESS in 3m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/18e3f6ff27ff4bf78fcd46ed7df7b6d5 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fcbb8bd5854246988bae5f3a0ea4ca76 : SUCCESS in 7m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54a291056d414baa95e111a133bf42e2 : SUCCESS in 41m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8334ef073a3a43efaa0f65c27d3eabdb : FAILURE in 52m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/13e4823e47284e55a6c577c77f627062 : SUCCESS in 1h 57m 32s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/718a8a1cb25c40bb881e3518977e9eca : SUCCESS in 20m 04s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/131b642dfce3429ab47ca2e01de3b256 : SUCCESS in 38m 26s","accounts_in_message":[],"_revision_number":2},{"id":"ba386b9b39277fb1461903b2ebd151f75bcf1c97","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 15:14:42.000000000","message":"Patch Set 2:\n\nrecheck\n\nError restarting container 8ad3f7cc0d5e514a1d4aedd722f3f955d0b7b202abae0c36452e94c9d971b58f: 500 Server Error: Internal Server Error (\\\"Cannot restart container 8ad3f7cc0d5e514a1d4aedd722f3f955d0b7b202abae0c36452e94c9d971b58f: ttrpc: closed: unknown\\\")\"\n}","accounts_in_message":[],"_revision_number":2},{"id":"163e08ef2a1ef63e45ed499a7d05c74ad3b315c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 17:00:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/750a169028914069b49e90bbc158dfb5 : SUCCESS in 5m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2afd36d468fb4b26b151cb02511c6137 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a819a2375c5344ecaa95e3e8c45f7c48 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c0f578506f14499b692f73d6d30d638 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b02741eac8e94e708dc4e20efce936d9 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c24af6a3d14295b13bd5f2470fb380 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95cd3e5534ec4c48bc330638700f147f : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/273dfa1d62914f768349cb8ff7f0c0f4 : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/72c71c0308da4c3095b60a4b9b26764e : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e0c87b9de434d15b6839fa690c878f2 : SUCCESS in 30m 16s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/de4507c7b9ee4a64991f0285f19110f6 : SUCCESS in 1h 04m 21s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/369ee5b8a86b4e6e8365ac3eb30be7dd : SUCCESS in 1h 42m 50s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2467c8fb378048549b8a57ccb1da2593 : SUCCESS in 19m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/9a2ab3c46fab4e458a76dc182037712f : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":2},{"id":"701b3cd36320f5be5956a626785fd67ee4a86337","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 15:18:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b1887ef1700f7cb49241e4306f659eeadfee8bbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 15:18:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b567bc84d0b8b28a0bafc9eae516ec38c6fe7840","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:04:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/63136ea81be04106885226705717aeca : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07bb335698f94a79a2d12f18708ca242 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a01d9d9ddffe45788588d54cdd927247 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3835413cee124e369b471e6c2be2be19 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb32f66482244f15a17d4d9d31c53067 : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc7637008cb14ef58a67ae4825d4bb19 : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f9ce11a9bc994f1e8b83062ac5fefbd1 : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c72dc592b22d46799261059aef6fcf4a : SUCCESS in 8m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dbb700accd39484c8bd7e554226a6e5d : SUCCESS in 39m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/280bb0ce02714e58881612785e426a91 : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/76ed03f88d854abf897ff77f2c10567d : SUCCESS in 1h 36m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ec0d3cd21b9d4951890003677eac78c4 : SUCCESS in 22m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a25aa8ebb080491e9f745d9857665a18 : SUCCESS in 39m 47s","accounts_in_message":[],"_revision_number":2},{"id":"d742b68e179ef11813ffe93f336f625346158481","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:04:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"21836ddf902597d0b294f122ddd78714c1bdde8f","revisions":{"3455e55973eb9eda5af5d1162549a1379fcafe9a":{"kind":"REWORK","_number":1,"created":"2019-10-09 08:55:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/687499/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/687499/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/687499/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/687499/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/687499/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/687499/1"}}},"commit":{"parents":[{"commit":"c620dd84f795a456bba3c9c8f916500e06846654","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c620dd84f795a456bba3c9c8f916500e06846654"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-09 08:55:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-09 08:55:12.000000000","tz":120},"subject":"Remove sphinx from test-requirements.txt","message":"Remove sphinx from test-requirements.txt\n\nDocumentation build environments use doc/requirements.txt where sphinx\nis already present.\n\nChange-Id: I3a2c9ae392745e001c5bdd01212c996e54c44f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3455e55973eb9eda5af5d1162549a1379fcafe9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3455e55973eb9eda5af5d1162549a1379fcafe9a"}]},"branch":"refs/heads/master"},"21836ddf902597d0b294f122ddd78714c1bdde8f":{"kind":"REWORK","_number":2,"created":"2019-10-21 09:10:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/99/687499/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/687499/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/687499/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/687499/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/687499/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/687499/2"}}},"commit":{"parents":[{"commit":"f9d0c256c7e03848cd5e7b0088cc7cc53fb6487f","subject":"Merge \"Try running `docker info` again if it fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9d0c256c7e03848cd5e7b0088cc7cc53fb6487f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-21 09:10:13.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-21 09:10:13.000000000","tz":120},"subject":"Remove sphinx from test-requirements.txt","message":"Remove sphinx from test-requirements.txt\n\nDocumentation build environments use doc/requirements.txt where sphinx\nis already present. Use doc/requirements.txt in the pep8 environment\nbecause sphinx is required by sphinx8.\n\nChange-Id: I3a2c9ae392745e001c5bdd01212c996e54c44f0c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21836ddf902597d0b294f122ddd78714c1bdde8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/21836ddf902597d0b294f122ddd78714c1bdde8f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
