)]}'
{"id":"openstack%2Fkayobe~688359","triplet_id":"openstack%2Fkayobe~master~I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05","project":"openstack/kayobe","branch":"master","topic":"dual2","hashtags":[],"change_id":"I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05","subject":"Fix issue with image download when checksum url is none","status":"MERGED","created":"2019-10-14 09:21:13.000000000","updated":"2019-11-04 14:29:11.000000000","submitted":"2019-10-25 16:02:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688359-1572019374918-b61b701d","meta_rev_id":"d1521342a5b6b6abf3e155010ffdad71057ca396","_number":688359,"virtual_id_number":688359,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 16:02:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-14 15:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-15 09:59:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-15 09:59:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 15:14:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-15 09:59:28.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-10-25 16:02:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c3568988a7b180f8372afdc51642a6532b47dfd8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-14 09:21:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b44a9d07781047f4b94146827c291ac9ca046b9e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-14 09:24:08.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"58a93b59fe6fec92e81ad14a9f151845f2504d99","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-14 09:47:06.000000000","message":"Patch Set 2:\n\nCould use a little story.","accounts_in_message":[],"_revision_number":2},{"id":"5e59d15b33b5ab2d27dce96e3535efd31e85eb1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 11:30:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75b532175ffd458298f9be2dbb1131ef : SUCCESS in 4m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a2c7ada581642469347a3c660a77af5 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76f90f8c585f452886546e3fb4ad1df1 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f0d4f36e3794bd7bf8ef67657776c04 : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2f97bfa8eb7f405f9d6a8bf9a63e5772 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06d6d82956344118a12b9be1eaf942f : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3dfe19e7f4b74f4db766daaa9410820a : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/396fbc4af3a342aebfcdabd704ee1b4d : SUCCESS in 8m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70e3e60c483e4a8a96bd2f1a7337b91d : SUCCESS in 26m 46s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2608a9b539a54602be1afd1622560092 : SUCCESS in 57m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d0ceda57c4c2466a89e2142baf959b95 : FAILURE in 2h 02m 54s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0fc27ab2b73f4664b00a8900633581e6 : SUCCESS in 16m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/2f9c47727a3a42f4b21e913cc8676398 : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":2},{"id":"6f04a179cd52d4fabe795c27e1a386c5816da116","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-14 12:25:05.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"15c9134b1088be0c200a14a87b24889c4803803d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 14:25:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/395714ee3d9444138ed81f6abbbf4a00 : SUCCESS in 4m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6529de3a63f1453988879ad91a87d4ac : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad100aa219cc4e679df941c3ddcb74fe : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c36153d436a443c9eabead23d201cb2 : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff9f7e11247f4268a771c975185eb3a8 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/961fdd3f1d734d31b831e7637a174b40 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c7a29450d1c04babbd44d7b1eb32cd7b : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca053a243ed64b519d2ac45cb43333ee : SUCCESS in 9m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e49932bc8ecd430297caaf104a073cfb : SUCCESS in 36m 30s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1a69bd14149b4e6888279f9418d02b3e : SUCCESS in 1h 06m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/752a73836d5d4e868aa98d0eda525762 : SUCCESS in 1h 55m 53s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/93a3bfb3ff5542bab983b0b4802d429d : SUCCESS in 19m 54s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/34f6c74a8e2644af9e8b1ccb548923d2 : SUCCESS in 41m 04s","accounts_in_message":[],"_revision_number":3},{"id":"eb02736ef464feb2b77ca501d143c2c521600441","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-14 15:14:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"8df764a79b63ec5eeca1c869f63041630d288a90","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-15 09:59:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"aad522dd469619a8319a94f08727b080501acb30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 09:59:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c29a583107923b3d9db6beaa03e26d5f3215e15e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 10:40:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/49e404b257484399a6c10d0ec428c130 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b99bf730ca94431a154093691f95458 : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e3a7a6e37da4c27a2b3f5c348f3af73 : SUCCESS in 3m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8bc5f2386074a02adcf27a48cdba450 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c88fd4a420c4cd6b6a9c9df022959c6 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c4a6c1fb9ded4d4b9ab2b358976f29c8 : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d201e1399f3c4ed29f4d8aaffb1b1c09 : SUCCESS in 7m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7d8caa5c230c44fc87264c5bcf9a6e48 : SUCCESS in 35m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c20a316a699b4073abf75960634ebfcd : RETRY_LIMIT in 3m 43s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d65375ab40864f769a357ebc3e4abdd0 : FAILURE in 3m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5e07c30c6ef54771876e2b3ddf0713fa : RETRY_LIMIT in 3m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/1f7ada061a9942e1810c3c599c27441d : FAILURE in 4m 06s","accounts_in_message":[],"_revision_number":3},{"id":"99b695dcc18f1a238dd2c284709bc8753ac98037","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 11:18:25.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"2a9ecb9c740c19c8d4ee3b13e7585a626688070c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:52:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e78e115993b43aabf173d14d8df84bd : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2a878c6332134a79b726d2e657903c6b : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17d3d5fa9dd94046bab2bea4c9973bd9 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec9c626fda0242e9ad9f7fe457bc1e12 : SUCCESS in 5m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b4dfe5457c6446898edeebe587662b7c : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e601b54b6edc4c91932f0288ac3aac90 : SUCCESS in 6m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ebfc0e838f2b495882039d3ed5a77781 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cf768ea0b32b43b19dd3f74573792179 : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/957f4dd273f0494ca28feb1e2c8d2a27 : SUCCESS in 39m 09s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9908f75a66194ff992fda07511421eb2 : SUCCESS in 1h 13m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/36cce5f1adf9498ebaf7de975dbd6e11 : SUCCESS in 2h 27m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6ab0694c553d4ce1929f05c545b2f0ac : SUCCESS in 21m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/167aafa2a3644d6ab67789d00286f05b : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":3},{"id":"398a40c802c4b68109d61a9d065fedf8cdf3f442","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:52:28.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"426854f75cd8751e14c85bec6cb45eb0fddf64a2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:02:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"156b737f8c1b38a04e0a15c51c326c1561e1357b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:02:54.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bf6ba56f641143ac8a027147965f5b96 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85b7c2a5aaea4d56967bd236ae5f3cb1 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41c5c0607cad430f9e04c6da0c947b80 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bfedeffa0a96455f88e96921535441f0 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc78578ee7eb478fa1245743f7d056ba : SUCCESS in 8m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b99deee3adf4fa0ba35e67c71244dcf : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8f2dfc640ccf48618810c4d2c45f5068 : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e370790d9780435fb376bf64b9e1bcf0 : SUCCESS in 40m 01s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e652dec6526a4ca0bd11a6903c2a63c1 : SUCCESS in 1h 20m 02s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f38a4a51a4b44583a3df826b598c0d52 : SUCCESS in 2h 04m 22s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e942112de72e406f935895b77fc02fee : SUCCESS in 19m 29s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d05006b04232417cb54bab6204fd24f0 : SUCCESS in 42m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a0d0193b934dbf73f731fec9c4a1f453ea679e94","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:04:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bd79a28dc8449c09e6c22992becd326 : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":3},{"id":"e87ffa3ddba4be5691e1cbfce09128ab67be45b3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:28:44.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 8e1b92436137181d560a5f21570fb9e44114cfad","accounts_in_message":[],"_revision_number":3},{"id":"135674c879bfc6b31e8f784114e63af6c554ca81","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:28:59.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit d4ea80760c2dd8f1bcc4c8079e379f721f00d242","accounts_in_message":[],"_revision_number":3},{"id":"bc5369b4c2f64a68e6afa667eb03e572a1939ce1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:29:11.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/queens as commit b62ec97cd43428470f88491a835054321d8730d4","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dea36ac9d24c1e8cf19f0d27ee49be657d736da6","revisions":{"66ca5637de69d0737c7d6602a76befdadb2bac2d":{"kind":"REWORK","_number":1,"created":"2019-10-14 09:21:13.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/59/688359/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/688359/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/688359/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/688359/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/688359/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/688359/1"}}},"commit":{"parents":[{"commit":"c620dd84f795a456bba3c9c8f916500e06846654","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c620dd84f795a456bba3c9c8f916500e06846654"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 09:17:47.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 09:21:00.000000000","tz":60},"subject":"Fix issue with image download when checksum url is none","message":"Fix issue with image download when checksum url is none\n\nTASK [image-download : Ensure the image is downloaded] *************************\ntask path: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml:19\nfatal: [localhost]: FAILED! \u003d\u003e {\n    \"msg\": \"The field \u0027vars\u0027 has an invalid value, which includes an undefined variable. The error was: \u0027dict object\u0027 has no attribute \u0027content\u0027\\n\\nThe error appears to have been in \u0027/home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml\u0027: line 19, column 7, but may\\nbe elsewhere in the file depending on the exact syntax problem.\\n\\nThe offending line appears to be:\\n\\n\\n    - name: Ensure the image is downloaded\\n      ^ here\\n\"\n}\n\tto retry, use: --limit @/home/zuul/kayobe-venv/share/kayobe/ansible/kolla-openstack.retry\n\nPLAY RECAP *********************************************************************\ncontroller0                : ok\u003d3    changed\u003d0    unreachable\u003d0    failed\u003d0\nlocalhost                  : ok\u003d6    changed\u003d2    unreachable\u003d0    failed\u003d1\n\nChange-Id: I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66ca5637de69d0737c7d6602a76befdadb2bac2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66ca5637de69d0737c7d6602a76befdadb2bac2d"}]},"branch":"refs/heads/master"},"aa3b81a8ae5e839253ca9fafa10a84f9a826ee68":{"kind":"NO_CHANGE","_number":2,"created":"2019-10-14 09:24:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/59/688359/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/688359/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/688359/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/688359/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/688359/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/688359/2"}}},"commit":{"parents":[{"commit":"c620dd84f795a456bba3c9c8f916500e06846654","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c620dd84f795a456bba3c9c8f916500e06846654"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 09:17:47.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 09:23:58.000000000","tz":60},"subject":"Fix issue with image download when checksum url is none","message":"Fix issue with image download when checksum url is none\n\nTASK [image-download : Ensure the image is downloaded] *************************\ntask path: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml:19\nfatal: [localhost]: FAILED! \u003d\u003e {\n    \"msg\": \"The field \u0027vars\u0027 has an invalid value, which includes an undefined variable. The error was: \u0027dict object\u0027 has no attribute \u0027content\u0027\\n\\nThe error appears to have been in \u0027/home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml\u0027: line 19, column 7, but may\\nbe elsewhere in the file depending on the exact syntax problem.\\n\\nThe offending line appears to be:\\n\\n\\n    - name: Ensure the image is downloaded\\n      ^ here\\n\"\n}\n\tto retry, use: --limit @/home/zuul/kayobe-venv/share/kayobe/ansible/kolla-openstack.retry\n\nPLAY RECAP *********************************************************************\ncontroller0                : ok\u003d3    changed\u003d0    unreachable\u003d0    failed\u003d0\nlocalhost                  : ok\u003d6    changed\u003d2    unreachable\u003d0    failed\u003d1\n\nChange-Id: I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa3b81a8ae5e839253ca9fafa10a84f9a826ee68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa3b81a8ae5e839253ca9fafa10a84f9a826ee68"}]},"branch":"refs/heads/master"},"dea36ac9d24c1e8cf19f0d27ee49be657d736da6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-10-14 12:25:05.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/59/688359/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/688359/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/688359/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/688359/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/688359/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/688359/3"}}},"commit":{"parents":[{"commit":"c620dd84f795a456bba3c9c8f916500e06846654","subject":"Blacklist sphinx 2.1.0 (autodoc bug)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c620dd84f795a456bba3c9c8f916500e06846654"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 09:17:47.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 12:24:44.000000000","tz":0},"subject":"Fix issue with image download when checksum url is none","message":"Fix issue with image download when checksum url is none\n\nTASK [image-download : Ensure the image is downloaded] *************************\ntask path: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml:19\nfatal: [localhost]: FAILED! \u003d\u003e {\n    \"msg\": \"The field \u0027vars\u0027 has an invalid value, which includes an undefined variable. The error was: \u0027dict object\u0027 has no attribute \u0027content\u0027\\n\\nThe error appears to have been in \u0027/home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml\u0027: line 19, column 7, but may\\nbe elsewhere in the file depending on the exact syntax problem.\\n\\nThe offending line appears to be:\\n\\n\\n    - name: Ensure the image is downloaded\\n      ^ here\\n\"\n}\n\tto retry, use: --limit @/home/zuul/kayobe-venv/share/kayobe/ansible/kolla-openstack.retry\n\nPLAY RECAP *********************************************************************\ncontroller0                : ok\u003d3    changed\u003d0    unreachable\u003d0    failed\u003d0\nlocalhost                  : ok\u003d6    changed\u003d2    unreachable\u003d0    failed\u003d1\n\nChange-Id: I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05\nStory: 2006715\nTask: 37086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dea36ac9d24c1e8cf19f0d27ee49be657d736da6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dea36ac9d24c1e8cf19f0d27ee49be657d736da6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
