)]}'
{"id":"openstack%2Fkayobe~688403","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Idc96523a5108712a072b14889b1fba0f05c381e8","project":"openstack/kayobe","branch":"stable/stein","topic":"dual1-stable/stein","hashtags":[],"change_id":"Idc96523a5108712a072b14889b1fba0f05c381e8","subject":"format pxe_append_params as string and not a list","status":"MERGED","created":"2019-10-14 13:24:57.000000000","updated":"2019-10-14 18:12:10.000000000","submitted":"2019-10-14 17:51:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688403-1571075462137-d762e54f","meta_rev_id":"c82211aea5b4338df77443abe11b967d3683f528","_number":688403,"virtual_id_number":688403,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 17:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"all":[{"value":1,"date":"2019-10-14 14:15:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":2,"date":"2019-10-14 15:04:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-14 15:01:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":1,"date":"2019-10-14 15:04:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 14:15:01.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2019-10-14 15:01:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-10-14 15:04:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-14 17:51:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53b29fe93eb82912fc54172043f5ce6e773b8727","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-14 13:24:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5af406b8f7cb2048359bb7d1a2b8d5da363c1bc6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-14 13:25:25.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 95acef72bc40e03cca7b6aac6b2428daa960b48b","accounts_in_message":[],"_revision_number":1},{"id":"b8ca79095e2c79edc0d3c015c56414a50e96d341","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-10-14 14:15:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"560a7c9d705f944285c41fffb0430dbc78826dec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-14 15:01:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45d9db5bccd404a11e78786a22be3bd99a47b986","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-14 15:04:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"36a66f4ebe053be999f5afe10f66cec92a2054ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 15:46:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e65b993102a44d0d9a07ea2637d610f9 : SUCCESS in 4m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7a945542f0784245af0c8f9adc66299f : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe48099d578a4adca1ec7f9b67c98ba2 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40d290c8854b47f0bfcfb50b7bf05f43 : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/70554faef12b49219d5077743e548c84 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a83565247ce4636a0f2923410d46332 : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/282aa238b85c427b8d8e38639717cdc4 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0de4868f8e41467fb7dc5d42d7ccf548 : SUCCESS in 7m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3c24551f33984558b331d6a3392166da : SUCCESS in 35m 24s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/64ba2f5823a64b038971391581bbef3c : SUCCESS in 1h 07m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/bd7dffd83bfc47bfaede7c8a26bc6dd1 : SUCCESS in 2h 18m 22s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1cbb329f865344299413ada41ad04c5e : SUCCESS in 17m 29s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/cac78660f45e4091a1682fa09007db0c : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":1},{"id":"f7bae79b9f667c0a7f4ba247e7980d2c34a9b4fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 15:47:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"02d81dabc891989ad49b9f4edf2b3b8af46da281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:51:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7e8dc71acb8f466b870563fa7953b60e : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da3941373ab2434c8f08b2cf3f978bd1 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/442964e9352c4007bc349295b8cde9ed : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9f108c0238142f780a76457cfdf8f4e : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07c6b02838aa436fb32550cc14fbf163 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a1fb5529a884ef5aa655904a81f9721 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb9c4ce6a0b94ce095147319f9e8c0e6 : SUCCESS in 7m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b0bea0068e94ca2991ffd82e7625e12 : SUCCESS in 40m 34s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9377a96f77a1484f81aba25673dee623 : SUCCESS in 1h 08m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/dd4ca2efe85440c8be76b5b22e7a9179 : SUCCESS in 1h 58m 36s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5ecff9a01ca142a09de81c5b03d3f19c : SUCCESS in 21m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3c1166aa96cb4ac7931c25d197645863 : SUCCESS in 40m 23s","accounts_in_message":[],"_revision_number":1},{"id":"80b587a52080c36ece2ac37f7ee6ebdeb50b87f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:51:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"61d0d700571a6fd116c4e98e6d5e9fc14c268e55","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:12:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/085a262987df4f259dda769c25b79b8a : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f2845f855e9bc1155df52d20b928d84b6370b94","revisions":{"4f2845f855e9bc1155df52d20b928d84b6370b94":{"kind":"REWORK","_number":1,"created":"2019-10-14 13:24:57.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/03/688403/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/688403/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/688403/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/688403/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/688403/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/688403/1"}}},"commit":{"parents":[{"commit":"f34154dad2c598998addf1fb671399dd9c977fdb","subject":"Fix container image build with multiple regexes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f34154dad2c598998addf1fb671399dd9c977fdb"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-11 10:26:49.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 13:24:57.000000000","tz":0},"subject":"format pxe_append_params as string and not a list","message":"format pxe_append_params as string and not a list\n\nPlease see: https://docs.openstack.org/ironic/latest/configuration/sample-config.html\n\nChange-Id: Idc96523a5108712a072b14889b1fba0f05c381e8\nStory: 2006707\nTask: 37067\n(cherry picked from commit 6430ec5261da83d4e4ad05cd55decede427e3856)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f2845f855e9bc1155df52d20b928d84b6370b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4f2845f855e9bc1155df52d20b928d84b6370b94"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
