)]}'
{"id":"openstack%2Fkayobe~688697","triplet_id":"openstack%2Fkayobe~master~I1bd724be7dc77058870f37cb1c9404472fa466ca","project":"openstack/kayobe","branch":"master","topic":"libffi","hashtags":[],"change_id":"I1bd724be7dc77058870f37cb1c9404472fa466ca","subject":"Install libffi headers","status":"MERGED","created":"2019-10-15 12:36:20.000000000","updated":"2019-10-15 16:26:52.000000000","submitted":"2019-10-15 16:25:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688697-1571156705654-01385176","meta_rev_id":"5b4c12d75130567ddfe9a3b6cbe516836e4ea8a6","_number":688697,"virtual_id_number":688697,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-15 16:25:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-15 12:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-10-15 14:34:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-15 14:34:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-15 16:25:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0646ac5e677c7c93cc9d8afe344df240c6493361","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-15 12:36:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fcadf7b6289fda448e66040a10a3a04bc89c67a7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 12:46:41.000000000","message":"Patch Set 1: Code-Review+2\n\nWe should probably have a bindep.txt file to do this \u0027properly\u0027. I like that this works in a dev environment though.","accounts_in_message":[],"_revision_number":1},{"id":"9b7c561713771eade4eb1935a25f4f6651bb4ddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 14:19:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f73045f0ec14ba2ae391357f652338b : SUCCESS in 3m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bef7067cbe70478d8c8a642a588f6db1 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6481453adec4fac8cc9a22eae77c4d2 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6494821371cf4ab8b0369046895f3e13 : SUCCESS in 5m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/185a3a63dff149b097e37b6ee2fc6be9 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e926d71a1adf48389575071f1ca02169 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/05c7f12d829e423db4afd427a27c34cf : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b92cf49ae6eb4b909bc517acd54ebd91 : SUCCESS in 8m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/07928923e2e74daaad6e5a45a6fe9d1f : SUCCESS in 30m 00s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2e7463ef0ea74077991706aae80be5e0 : SUCCESS in 1h 08m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d12ccb44067946b390914a7676e96cc1 : SUCCESS in 1h 39m 39s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/86b77d0bc07d462390ea3701442974a7 : SUCCESS in 21m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/2225f51df646447eb7d0ce561b0b2a81 : SUCCESS in 44m 11s","accounts_in_message":[],"_revision_number":1},{"id":"7cbc042a78fa817ded9df8675e674ec471210274","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 14:34:50.000000000","message":"Patch Set 1: Workflow+1\n\nUnborking CI","accounts_in_message":[],"_revision_number":1},{"id":"999f2b26379537bd7bca128aac8774971d396750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 14:35:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"04dc87db10a05be410c8959e331787acf9d7afba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:25:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0cb0b1b3bca953661fe39a1582689efbe54d3035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:25:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/13ee7c68179947259d60391e840e55da : SUCCESS in 3m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2db4c3551b7444ceae9e0f3bd4ae5f65 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef4c0c4af7c4d108dbebc4eb4bb6808 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a3c244f9f154a059f45ba24a38c5484 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab66c829d404008bcd684fdc2e3ae8c : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62e4b5664f0d44568960db14848b3ff4 : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0debedaf93ca4c2c89beae6cc3d3c838 : SUCCESS in 7m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e12fd83bf5744db08274fb916946b136 : SUCCESS in 37m 02s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e75d86eb618f495fa33a170f6ab27fe8 : SUCCESS in 1h 20m 14s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a7a852ad472445ac9e7c5ba26a5b9f42 : SUCCESS in 1h 45m 04s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/14e95b4eaa13456c9ddda0eefe7ac1c6 : SUCCESS in 20m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c9c1b0dbe1434fa2a548ab80ba7c58cd : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":1},{"id":"58de6c783ab2f552544402ebe89c4757a7ea6480","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 16:26:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdb3851bb8484d99b3307db9f8afec58 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3d3649c5ef555bfc73786258030a6986e779cc9","revisions":{"a3d3649c5ef555bfc73786258030a6986e779cc9":{"kind":"REWORK","_number":1,"created":"2019-10-15 12:36:20.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/97/688697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/688697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/688697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/688697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/688697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/688697/1"}}},"commit":{"parents":[{"commit":"6430ec5261da83d4e4ad05cd55decede427e3856","subject":"format pxe_append_params as string and not a list","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6430ec5261da83d4e4ad05cd55decede427e3856"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-15 12:35:15.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-15 12:35:15.000000000","tz":60},"subject":"Install libffi headers","message":"Install libffi headers\n\nThis is to resolve the following issue in CI:\n\n    c/_cffi_backend.c:15:17: fatal error: ffi.h: No such file or directory\n     #include \u003cffi.h\u003e\n                     ^\n    compilation terminated.\n    error: command \u0027gcc\u0027 failed with exit status 1\n    ----------------------------------------\nERROR: Command errored out with exit status 1: /home/zuul/kayobe-venv/bin/python -u -c \u0027import sys, setuptools, tokenize; sys.argv[0] \u003d \u0027\"\u0027\"\u0027/tmp/pip-install-PV3WhJ/cffi/setup.py\u0027\"\u0027\"\u0027; __file__\u003d\u0027\"\u0027\"\u0027/tmp/pip-install-PV3WhJ/cffi/setup.py\u0027\"\u0027\"\u0027;f\u003dgetattr(tokenize, \u0027\"\u0027\"\u0027open\u0027\"\u0027\"\u0027, open)(__file__);code\u003df.read().replace(\u0027\"\u0027\"\u0027\\r\\n\u0027\"\u0027\"\u0027, \u0027\"\u0027\"\u0027\\n\u0027\"\u0027\"\u0027);f.close();exec(compile(code, __file__, \u0027\"\u0027\"\u0027exec\u0027\"\u0027\"\u0027))\u0027 install --record /tmp/pip-record-ZvlZVY/install-record.txt --single-version-externally-managed --compile --install-headers /home/zuul/kayobe-venv/include/site/python2.7/cffi Check the logs for full command output.\n\nChange-Id: I1bd724be7dc77058870f37cb1c9404472fa466ca\nsee: https://zuul.opendev.org/t/openstack/build/c20a316a699b4073abf75960634ebfcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3d3649c5ef555bfc73786258030a6986e779cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3d3649c5ef555bfc73786258030a6986e779cc9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
