)]}'
{"id":"openstack%2Fkayobe~688725","triplet_id":"openstack%2Fkayobe~master~I65e0175402441ba6422f12ebd4baa06640b9181b","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I65e0175402441ba6422f12ebd4baa06640b9181b","subject":"Add nova-super-conductor to inventory","status":"MERGED","created":"2019-10-15 14:45:51.000000000","updated":"2019-10-24 13:25:01.000000000","submitted":"2019-10-24 13:23:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688725-1571923418614-fee2ba80","meta_rev_id":"e62d22b95921cb15427220d4bd597d936ba5a9af","_number":688725,"virtual_id_number":688725,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-24 13:23:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-10-23 12:52:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-23 14:19:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2019-10-24 09:14:47.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-23 14:19:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-10-23 12:47:37.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-23 12:47:37.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-10-23 12:52:26.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-10-23 14:19:37.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-10-24 13:23:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c0c24e9926054ae9025945711dfe1d3d30aa306","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 14:45:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e367c1b617b6d0b7ffbc26c48339add8c75101d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 14:46:09.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"3ad38bb0ca759ca642fd81730c85661d690b7916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 15:55:17.000000000","message":"Patch Set 1: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"6d90df54bd95f293e4a881c5606491ceda222be7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-15 15:58:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e4adda76e646f1d7f48e4d9508454f79c795725a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-15 17:05:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c7a6bc926a24f4ca173e632c5d93fc0 : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/93e68f2e810b4cbbb57dc29881019773 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7503199ced44767a3cb893edc171f60 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76f649c4932b4593822ae4377bc175f9 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8d6d0c845d13454fa3472dc9434e5957 : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b429ecf061a646ca9c1b380e5d52570b : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/42140ceadabc4032895020d6884f5930 : SUCCESS in 3m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9f8bed37945446ecb0e8f5c787c4226c : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/89b077e917674c6b8d6f06a812a01a33 : SUCCESS in 38m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/026e627cd9c44e81a5872f44e58031e3 : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2b1dbcd1d38b49d3b1922713a7876398 : FAILURE in 29m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/57445b41bbfe43e4ba8940edd391fc8d : SUCCESS in 19m 54s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ee2c49c1104f46beaea21416e2374290 : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e017a0ef7b354ab14f9aff605eae3e3b036bd3c7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-16 08:53:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"27847f053e332d3f006fdbdc4e530b8db777e54b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 10:58:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f3f0c2bffced41ffaa91d032682c25b3 : SUCCESS in 4m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ace6f58625514a349a6394c37389aa38 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a95cb5d540e04e9ab240d36eec2467fd : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da4b7d1abaf74aa29836f22e1ced3610 : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f2868f447bde4fa7a93879d695841679 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/946f5a86490d41b8a27300541ad46d79 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61e9c483620849c2a3c3e99253f090b5 : SUCCESS in 3m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3ced8ccc1dd14f8991e48de2bacdda8a : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/331d75abce854c4cb695b9db04b0b870 : SUCCESS in 35m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1d6ff591610f472aaa3b6c64bfc6ea25 : SUCCESS in 1h 10m 34s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/99ad0089d5ee4aeb83b2cd9d9f0f2a28 : SUCCESS in 1h 57m 00s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5578070821de413eb5c2d0932fd3258a : SUCCESS in 17m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/104042ab38ee439c8d7434d1ff7697eb : SUCCESS in 35m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cbdfeb3afe05cedf8269eb9ae9db6cc4be87d1dd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 12:47:22.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"6339e3faed4640af920ad084bc6fb6c29ae6e748","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 12:47:29.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"15198df571c020ff9ea811efcdfde98edbff5a2c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-10-23 12:52:26.000000000","message":"Patch Set 1: Code-Review+2\n\nAdd a release note to help advertise the fact that Kayobe supports cells v2?","accounts_in_message":[],"_revision_number":1},{"id":"68f04bbc435e4318096c4e7c10824c1fd60cdc3d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 13:02:09.000000000","message":"Patch Set 1:\n\n\u003e Add a release note to help advertise the fact that Kayobe supports\n \u003e cells v2?\n\nWe really need the group_vars support for that. This is just to get CI passing again.","accounts_in_message":[],"_revision_number":1},{"id":"6d5a826ca212a44cf11ad28970a97d075923dc4c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-23 14:19:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nUnbreaks CI","accounts_in_message":[],"_revision_number":1},{"id":"8f42b2987823dd0a919de19c75d9a4ccaa6dbeea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:19:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d06dc5c9cfa174716efb151d2312578aab376be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 15:15:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc98a839f96b4573b5648abb42f08f37 : SUCCESS in 3m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/634f8a481e1d4ed28ecb9db4fd2b0bdf : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e98d7dd593ef4a928b8a342cc40c6666 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a1a81a0c3e041ac926a5e562bb74b5a : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4cb186a70efb4d979638507fa2ab5487 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52bc4b096c744545acd3521f91b61de7 : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b853443805d84f14b9d15a195fc2f6bc : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/838a0a2aafdd42c49fe0ead424e3f2af : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f98e6ddca69743a1be4229d3edb6f12c : SUCCESS in 27m 06s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/221cfd6d503845efa1235657a769029e : SUCCESS in 1h 11m 19s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/dcf9da9eb7e347218642255943137113 : FAILURE in 2h 12m 45s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f42ab44c841a40e8928aa9b3beba6a7c : SUCCESS in 24m 51s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f1adc068956e45e6b6c4cb1df3697247 : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a9e67a39c3d1d89eff8b472f74b9c5986b43a0ae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-23 15:47:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"75ba214aee58c595fbebbad0f488357727092dec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 16:24:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c252f51994c044f29d07ea762cfb0ce8 : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5d4a4eb3bf94c3a95fb7536fa830a38 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd2ece1688ce4ed0b53c41e2ba5ed2ad : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/631e563308fe4412aa9273b7d4b0ae33 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfaa01fd5f0e45b2b8708451ca488dd8 : SUCCESS in 7m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f30f1a4fcb1479a8687d82dd5122d99 : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ee2b887977847dea3895908df382f09 : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9251b26dc9374480a3cd1ed87c3cd9af : SUCCESS in 37m 58s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/05c7cf8f85ea4934a6ea459a5558408f : TIMED_OUT in 2h 02m 03s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1badd687e5b94dcb8fa8100bd12821b2 : SUCCESS in 1h 51m 57s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/39e9d4f0559f4593919d550849445dcd : SUCCESS in 25m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8979ff96a29d47a8b103b2c6aedb6a26 : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":1},{"id":"4b0fb953eebf87fbc6fe9c074e26bd0924655492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 17:56:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/557bf9e157f64796951298a4f22d0b50 : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/da53a96a80e445bda83fc5f0d7d766ff : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/251dd7112c9d42338864d00baced9333 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c21913b016479d8597494b0866f7b5 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7770c4e7a4c4be89c79d196d1915457 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/568deeeda50943cc8a80d7239ce967eb : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d8a5855e252a490e9467ae70b00ab4d5 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8e7ce98bea24660a8b7cf7e0a999203 : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c79ea04d1e524fba87ed7770bc2187cc : SUCCESS in 39m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1c5cd53e6ed74604a5d585f3673ebcbb : TIMED_OUT in 2h 02m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c06aa4ddf82f42aebf29828acddb5f4f : SUCCESS in 2h 02m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/d6beb231438b4982b7b13b9c08328181 : SUCCESS in 20m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/37ab2b14b12e41fba2a2fab8cd9151a0 : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":1},{"id":"8216ea431832c3cb40f562fdf9ca9f296ccf0431","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-24 09:14:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3b4fb2f1deb93da89d2f27c3987bfd91308e47f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 11:15:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ad3025c82eb4d98bbea47c42df079c7 : SUCCESS in 4m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/be5ad218702842c093f400937ee1c6c8 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5913ed7b9e32419b9ee7dfaf955996d9 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7e9db92621427ea1bfed7f8a982da4 : SUCCESS in 4m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c22f16151004394813632019a3553a5 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed8dd20274b84e9abc23dc1fe7e69c1e : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d3df61d4f996433cad9f2adc07f9076c : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35e235d699724bd8a63c55b9480b0c10 : SUCCESS in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4dfd3754de041a9b04f2341f2df4658 : SUCCESS in 38m 51s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/47a38403938b4322aabe84df826cf26e : SUCCESS in 1h 14m 16s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/de02d57ec1f74168af1c719a5dadce10 : SUCCESS in 1h 55m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b7a63034983a4b6d9b6c2fe7ae2fb18b : SUCCESS in 22m 53s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4b5b9b646a194142bb41a741158009bc : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":1},{"id":"a98e5c7b02c1057999ea2ec7bafa7ef23824f29f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 11:15:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dfb99f18c0b524f9f235f49382f2ca18e8e16ad8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:23:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"24f300dbc8e9d14f04c5a1545b55680ea8f07968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:23:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6f4d07b3b3c34d1e95516a4e31e82a8d : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65da2d49458f4e8cbbfb76f933a2b699 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d54cd4b0daf4d1c83780d7bb42c59ae : SUCCESS in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cd8a9a512c764a74a756441d20deb573 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cf519a8473b420eafec07d6b1fb63e4 : SUCCESS in 5m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00c04e3a213342679b83b064aa162b03 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7e59a7e7919848cb9f012e16dbdd07f9 : SUCCESS in 7m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee309942ea5f4b9996fb4af137b9bd5b : SUCCESS in 37m 03s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ba86ddd41dc343f3b7fb031729eecfd1 : SUCCESS in 1h 21m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/dbfd3f73236645fd8e16b226eb800dea : SUCCESS in 2h 02m 03s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/90574a1e22d64b92a523b06edf2b1aef : SUCCESS in 23m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bf5b66c9b13441c2b5f2690e73b39540 : SUCCESS in 41m 18s","accounts_in_message":[],"_revision_number":1},{"id":"e62d22b95921cb15427220d4bd597d936ba5a9af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 13:25:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82bad07a1b0740619b72ab3705a4bba7 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56b98f93c9b07d11ee05d1fdc69c06b94b12878c","revisions":{"56b98f93c9b07d11ee05d1fdc69c06b94b12878c":{"kind":"REWORK","_number":1,"created":"2019-10-15 14:45:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/25/688725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/688725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/688725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/688725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/688725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/688725/1"}}},"commit":{"parents":[{"commit":"886fe4fd7694c365c13c0063fb017ce0425a6664","subject":"Merge \"CI: Remove rocky pin from molecule requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/886fe4fd7694c365c13c0063fb017ce0425a6664"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-15 14:42:26.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-15 14:42:26.000000000","tz":60},"subject":"Add nova-super-conductor to inventory","message":"Add nova-super-conductor to inventory\n\nThis is a new group, added to kolla ansible for multiple nova cells\nsupport.\n\nDepends-On: https://review.opendev.org/675659\nDepends-On: https://review.opendev.org/688697\n\nChange-Id: I65e0175402441ba6422f12ebd4baa06640b9181b\nStory: 2004291\nTask: 37132\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56b98f93c9b07d11ee05d1fdc69c06b94b12878c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56b98f93c9b07d11ee05d1fdc69c06b94b12878c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
