)]}'
{"id":"openstack%2Fkayobe~688990","triplet_id":"openstack%2Fkayobe~master~I24a8832c02db6ba019ab8f5c2b9d7216a9b7d213","project":"openstack/kayobe","branch":"master","topic":"libffi","hashtags":[],"change_id":"I24a8832c02db6ba019ab8f5c2b9d7216a9b7d213","subject":"CI: Don\u0027t set cpu mode","status":"MERGED","created":"2019-10-16 17:47:40.000000000","updated":"2019-10-25 16:12:45.000000000","submitted":"2019-10-25 10:41:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688990-1572000095717-256045c0","meta_rev_id":"9265c474747791da8006ed1933981b697176a93b","_number":688990,"virtual_id_number":688990,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 10:41:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-10-24 15:33:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-25 08:36:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-25 08:36:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-10-24 13:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 09:13:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-24 15:33:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-25 08:36:59.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-10-25 10:41:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1452f6f0bb023b4dac61e62080b9a4bcc9085ef8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-16 17:47:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74e03917a697ac8e490ad060d09104b3f8cbd112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 19:00:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2fd0c0d8b974c20849d2ab5996716e4 : SUCCESS in 4m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c5d142b5e99b4b898c8c087995807d13 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3702d9733c28492c80b32a08b5f0b9d1 : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c131958547bc409d9778c52b203f11c1 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7b19e4375f3440ad9a060711a9933062 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/452ae9e1985440708c00d13f7d8b51fd : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8fe8807f55e41168e69c49ce085b05a : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/08e823347b8c40b884f62b6cf6a3c4f6 : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/82351ca8426741bfba023bbcbaa37f63 : SUCCESS in 40m 57s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3b5dc9bc0b1e47239b9d0849c4fd9cc2 : FAILURE in 1h 09m 33s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5158f8f187fa4ee1a447219dedf1ccfd : FAILURE in 1h 00m 00s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/520e90aae0b8401788d76b81c5fc1724 : SUCCESS in 22m 54s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3554aeded395474b9b32cfbe0ffd4da2 : SUCCESS in 36m 44s","accounts_in_message":[],"_revision_number":1},{"id":"d8f33d637cea1214f110ffdb5049fcde470827f6","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-16 20:03:34.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b06da0dae7923ee22cecc7424e1b720fae8de581","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-16 22:09:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d30fa4bea37a40e59cd3e2ab2c62f23b : SUCCESS in 6m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1b30f2030d294ebd95f608d35ab08b31 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa47d6a5b6eb409bbd7776b461834d05 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072daf3cd5e74842b8c0ff3fca8ab411 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b60b4b939ea0408d9d484451fec0dadb : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e2a73db83c44a16a8f282bb8aa72713 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d28d78359f2443ab9ed2e554144d41f : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc1a469bebd4422bb68744e314733081 : SUCCESS in 9m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/491c711b033e4b27aba4c2afcec9660a : SUCCESS in 41m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/342a0e8fa00e41f0b631f56e70bc21a5 : SUCCESS in 1h 18m 19s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3f35bd40a81244e49be5f6551b7aea28 : SUCCESS in 1h 44m 30s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2f38e2a631dd4c98b5ac0fabdcc4e3d8 : SUCCESS in 18m 54s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d0c3631447b14f5d92e1d05c6efa22fa : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":1},{"id":"23b2d2e5bea82d8b49f8444a2034d6a9a0a6a0b2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-17 07:12:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ef04a5b58b13a8086d80607615e9002e5789c676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-17 09:23:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a61ac232372440da706530637e686e0 : SUCCESS in 3m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/026d7bc9e491439884f4dcd623806be3 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bea60ebe8ef24488b64a4623cc1c214e : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e7842e6a3cd4d0191826a71ebd3a691 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b573567b3d55427981201207d571e960 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a196a6cc0c4eebbfc23a02f795e897 : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dd6f8cf5d0194b2682cf9d08d8c3a432 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc1b60ddb17746d09222c63d1357b5e5 : SUCCESS in 7m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d9348f29f6bc4c00b7db677e569ef771 : SUCCESS in 35m 17s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bc394a9a17a54b25abc9557a9451fc69 : SUCCESS in 1h 08m 22s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0995b721865d4319b67ce627a142191e : SUCCESS in 1h 59m 05s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3352276fc1cb4216938f2d7ca5f7dccd : SUCCESS in 20m 35s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b522838b1b864845a47655d75b18e97c : SUCCESS in 38m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9fc76d2bf76dc049beb0947a508068737d15dfd6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 09:13:51.000000000","message":"Patch Set 1: Code-Review+1\n\nlooks like it helped","accounts_in_message":[],"_revision_number":1},{"id":"6bcc3716f5458dfee8ef51b082ef9f7c6fae6c15","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-21 10:28:36.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9c751b7428570abe10b4d83ece464ff6206d3814","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-21 10:29:15.000000000","message":"Patch Set 2:\n\nRemoved the depends-on. https://github.com/stackhpc/ansible-role-libvirt-vm/pull/32 has been merged and a release has been made.","accounts_in_message":[],"_revision_number":2},{"id":"8da03628e3611831188684023536884ea5aeed6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 12:01:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e11c8e3203a4a868a375a4783f51bff : SUCCESS in 3m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c76551115b814a758daa28cbae3fa47a : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/13d52615303c491f9635380bae5dcc3a : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/532522aefd7a4352ab57a80f89115318 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec6d77d308094509bbac3c486d7dce39 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6107e8cc342d473697f7153d5135bd8f : SUCCESS in 5m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2700f862bb84680a031d95c7082fc76 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cf3bcaa7d49a4c4e887ccc00a45bb382 : SUCCESS in 10m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/718abbde4e9e4148a576862415731a60 : TIMED_OUT in 1h 01m 24s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/47ace40a0e75432da39e771ec384a4a7 : FAILURE in 26m 22s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/028ca48c3c6141d0b70e0fa23fd817aa : FAILURE in 1h 26m 28s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/63668d1b254f4705a10cb8d604e308c8 : SUCCESS in 21m 59s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/90ff70c789344dbcaeb7fb420ac5326d : SUCCESS in 38m 31s","accounts_in_message":[],"_revision_number":2},{"id":"18c74400e7a76e51e8c40380106fe1d16c8e056a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-24 13:29:15.000000000","message":"Patch Set 2: Review-Priority+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f8b69db95ad5a746d8adc2392e12b4f0e57b3f77","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-24 14:37:32.000000000","message":"Patch Set 2:\n\nFailure was due to due to an issue with ansible-galaxy:\n\n  2019-10-24 14:12:11.100088 | primary | Running command: ansible-galaxy install --roles-path /home/zuul/kayobe-venv/share/kayobe/ansible/roles --role-file /home/zuul/kayobe-venv/share/kayobe/requirements.yml --force\n  2019-10-24 14:12:36.878462 | primary |  [ERROR]: failed to download the file: (\u0027The read operation timed out\u0027,)\n  2019-10-24 14:12:36.881316 | primary |  [WARNING]: - singleplatform-eng.users was NOT installed successfully.","accounts_in_message":[],"_revision_number":2},{"id":"03ba61095521b918b9eeaecf5a879a579e50cc40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 15:22:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79e9136b92174f79864fbe4652baab45 : SUCCESS in 4m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/952e1d28be2e48c9ae6043562d05a140 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8651f22285b4bc88d8a6b31b8af511c : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66928f894ca4478db3cae1e8bb9f73f0 : SUCCESS in 5m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bba574bb2814d75a6ce7f2b78bdd533 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9a569d25c0433fb29d60cf4f51e14a : SUCCESS in 5m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/81aa1b8c580346aab1c26e858afcc89c : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d24c2b1a2ca6476ea0cff0c531e34e93 : SUCCESS in 7m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/77142b1fa32a4162b1bd6261bdc7811b : SUCCESS in 37m 01s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e7ea289b38c24a95978152ebc6ce76c1 : SUCCESS in 1h 01m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0a4ad13000bb4a0794d3cf3b84da275c : SUCCESS in 1h 32m 54s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/78ea48c5c4b3463084ce360b68aed859 : SUCCESS in 23m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/504ce89104ea41a385a8139c16348597 : FAILURE in 23m 40s","accounts_in_message":[],"_revision_number":2},{"id":"49f80ed2dc853648060e166a2ce4b6c83d2f1aa0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-24 15:33:53.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"953a8125fa1e1013a965c9dc9666f9b962e5fac0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-24 15:33:57.000000000","message":"Patch Set 2: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"19af06bf96c17ec85b856a2d91c307d650a3588e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-24 19:57:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1dc320005cd14cf983b8a239ca8a5e81 : SUCCESS in 6m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0aba557f3db344a3b518def17220ce9a : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e67f9f6f7234193ae84c5b920d5eafc : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0c6282ccdd4763bb88c64a2d76f54e : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a1fc32dbfdd47858ae412b36d34ab57 : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae22eeab0d6a445aac6546fd979e0431 : SUCCESS in 5m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/58104a6e8c43463db38a274e720ce5a7 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/373f4e4873444c11a19066a54dd4d733 : SUCCESS in 7m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33fa2386b7db40a7acfc9a656a10da19 : SUCCESS in 43m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f8f2f688f1e644a4b9ada9f8d2e9ebf1 : SUCCESS in 1h 16m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/71fdae89285540e788a0eeafd77b15ff : SUCCESS in 2h 25m 48s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b415c7dcf3a44f169d2540a07dde56b0 : SUCCESS in 24m 06s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/88cee87a61cc46d2aeec66d9330c1619 : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b246f3d64825205c379c3ec04239f592ca4e23c0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-25 08:36:59.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nMaybe there is a better way to avoid this issue, since there are reports that this is fixed (https://bugzilla.redhat.com/show_bug.cgi?id\u003d1371617), but let\u0027s go with this for now.\n\nI don\u0027t think Zuul will keep the logs indefinitely, you may want to copy them to a paste or directly include the relevant lines.","accounts_in_message":[],"_revision_number":2},{"id":"ff14a6cc4b487bb28e7388519be7150f4bdd88d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 08:37:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1898d8606e9dcd2ecd2bf9a6f424b96bda8e231e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 10:41:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/15485a5bcfe24145a1d8a6d23f6ceb91 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/edc9aed0ec264cd99888f497d131a9b7 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95d8ad6bfa714fa5a5666cf747c7c115 : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/01e7c788727c441083d94d833b6574e4 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcc9de2028141d1822d6022c5a8602a : SUCCESS in 7m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dc16ff6776f54f3c9928f6c7b22b822d : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa178efde18e4c0abe3a3c3ecafb6b92 : SUCCESS in 9m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63c47f3323d149bd9f252c68c9b51aa4 : SUCCESS in 41m 23s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d25ff05875884dcba00160d55903196a : SUCCESS in 1h 45m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b31b0bba90664421b7406468c116a69b : SUCCESS in 1h 51m 45s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/bf5419911cac478bb6b5407adc50b709 : SUCCESS in 20m 40s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/629a6a07e1c24b68b0263c779ea58edc : SUCCESS in 35m 47s","accounts_in_message":[],"_revision_number":2},{"id":"893d3c514747af2efbb3223fc191fdeb0ab9407c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 10:41:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4231cf85b569c780350ba0df1e3b6ca468cee6ac","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 10:44:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/246438a72158412fa1053e43bb65b3ea : SUCCESS in 2m 00s","accounts_in_message":[],"_revision_number":2},{"id":"4f82c95464a3577aa93f501b92a37ac8ca66633c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-25 16:12:36.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 2d280086d511bc63d03a774761351f2ae71c1266","accounts_in_message":[],"_revision_number":2},{"id":"8dfb060550f622c4cccb49dc37a1b39e9532daba","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-25 16:12:45.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 660c2c55b8bf8c9b64f3434cedacb9fe58201cfd","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cc71fd03f9cd7096361efecc8e1eed6b0862eed2","revisions":{"40bb18534ed641b170511ad45196e2a1ea284f27":{"kind":"REWORK","_number":1,"created":"2019-10-16 17:47:40.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/90/688990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/688990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/688990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/688990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/688990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/688990/1"}}},"commit":{"parents":[{"commit":"a3d3649c5ef555bfc73786258030a6986e779cc9","subject":"Install libffi headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3d3649c5ef555bfc73786258030a6986e779cc9"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-16 13:52:57.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-16 17:46:39.000000000","tz":60},"subject":"CI: Don\u0027t set cpu mode","message":"CI: Don\u0027t set cpu mode\n\nNot setting cpu-mode will reduce the number of TCG emulated features.\nHopefully this will make CI more reliable.\n\nChange-Id: I24a8832c02db6ba019ab8f5c2b9d7216a9b7d213\nDepends-On: https://review.opendev.org/688975\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40bb18534ed641b170511ad45196e2a1ea284f27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40bb18534ed641b170511ad45196e2a1ea284f27"}]},"branch":"refs/heads/master"},"cc71fd03f9cd7096361efecc8e1eed6b0862eed2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-21 10:28:36.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/90/688990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/688990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/688990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/688990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/688990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/688990/2"}}},"commit":{"parents":[{"commit":"a3d3649c5ef555bfc73786258030a6986e779cc9","subject":"Install libffi headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3d3649c5ef555bfc73786258030a6986e779cc9"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-16 13:52:57.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-21 10:28:07.000000000","tz":0},"subject":"CI: Don\u0027t set cpu mode","message":"CI: Don\u0027t set cpu mode\n\nNot setting cpu-mode will reduce the number of TCG emulated features.\nHopefully this will make CI more reliable.\n\nChange-Id: I24a8832c02db6ba019ab8f5c2b9d7216a9b7d213\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc71fd03f9cd7096361efecc8e1eed6b0862eed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc71fd03f9cd7096361efecc8e1eed6b0862eed2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
