)]}'
{"id":"openstack%2Fkayobe~689709","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I5ae37b8f264437a9e49b09da459e604191f9ed3e","project":"openstack/kayobe","branch":"stable/rocky","topic":"docker-info-retry-stable/rocky","hashtags":[],"change_id":"I5ae37b8f264437a9e49b09da459e604191f9ed3e","subject":"Try running `docker info` again if it fails","status":"MERGED","created":"2019-10-21 10:24:38.000000000","updated":"2019-11-01 23:00:15.000000000","submitted":"2019-11-01 23:00:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689709-1572649215291-8fb7e297","meta_rev_id":"94d00050619a3f439cbc901bec78b513745a579b","_number":689709,"virtual_id_number":689709,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-01 23:00:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-21 12:52:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-21 10:25:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-11-01 14:52:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 10:25:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-01 23:00:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8c7d15a5f18f1d616d1a0d7d7961f92d16281465","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 10:24:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eeb6762f9818c2a96a820b5c279973a17c0dd212","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-21 10:25:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"87893002adad80149c1a7b8ded20147030f6afa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 12:29:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/0b2293c09a1e4ff6b97a21989801f849 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84dbe8a846a4e1a87093c60c89b5fc7 : SUCCESS in 3m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82edfc9eacc049368b35e47b4640e31e : SUCCESS in 4m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/378f0aef4f844db9be07e652ed768274 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4e7330cb67347aea1c674a5b5ab76e5 : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/48429d01d8cd45efa067e4b3457d5623 : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ece3e2ba41954029a77b440cafd3653b : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3abb05999bab40af999994beb0795712 : SUCCESS in 7m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c9c3b2595c84b8ab5d39fe5ab3dd656 : SUCCESS in 36m 50s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6c8efcb5e85d40ec8dfd174d329c3d57 : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/fd005348303d4e1a91884f7df70c866b : SUCCESS in 1h 52m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4ed9cb976dc248e286f958b4672b629c : SUCCESS in 22m 05s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/16d77408753140db8fef7308833a26c1 : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0b37270aed3793de59f9907856a989c017d8f12e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 12:52:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"10db12992cc74c13958fbf19a208ef39a7393cee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 12:52:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f8353dc306c9d5eeb592769d6901ecdba128cb6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 14:42:31.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7819d5cd2c9c40bd8ff5198c86fcd260 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d1f728131544d42b4470dd560ef19db : SUCCESS in 3m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c58260b3d67b499db7c07aba35a06dd6 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7593a06617e740ec8a6455d392cb8fe2 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/72d73f260bd247c5ad146958f8484b48 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f873d354a8f40ba9a50492587ce1e4e : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a491f44f055479aa6eb0df252c814e9 : SUCCESS in 7m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/580266e5e5344aed9a8f4df158c2c132 : SUCCESS in 30m 05s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f2ca108b07c14b5f86d7c7f9dd038b72 : SUCCESS in 1h 04m 38s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/965f0a6afb7643fd95d26920427e08b4 : SUCCESS in 1h 48m 43s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4661960ca0b54a098df3099d38b38c7d : SUCCESS in 20m 32s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4d47c81822fe4e688e2229f72e5242b8 : FAILURE in 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"acbc0a974cd19107fdb16c1b3004b8e391abf548","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 14:56:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"47e0af55f2e63063cbba1d5933652560b68cab82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 16:41:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2a4e8700e933415b9f36038919de9a71 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae12d1c57654950a7e88153e5ec02ca : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c5b966c044e430a9f276eca3fa10dfe : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2537ee6c3e924a3a86a0053131e7a378 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc9249e706af4aa094ec261461471c2b : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d45c772fd3204ab5b148164b3e4be50a : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9ba5480ce2ff4ba095bbc460c7beb08c : SUCCESS in 4m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/90ebeab7b59c4fd3a47bfed0c4896d71 : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/733decc0628e49f6a0caf1456a88e84f : SUCCESS in 33m 43s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e315c61f7d1940fda7821a36ebfc53f7 : SUCCESS in 1h 07m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/eac74f00e640481bb18e91811705b115 : SUCCESS in 1h 38m 20s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/810a53e297034408bbd497f6c8a1115f : SUCCESS in 19m 52s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ab8b3481b6e34fe19090341ced424bde : FAILURE in 21m 29s","accounts_in_message":[],"_revision_number":1},{"id":"c1ecda27dc5a24a3f7b971eff64d0a74c77b202b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 16:57:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ea166c879566ea5c6493284344907e05621b3af4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 18:47:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7d47e038e3994c28a657d1fccb6e4009 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a77a127b54542aa8dad68366994f61f : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9993164159e8483cbdef67bd954f6f2a : SUCCESS in 5m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ce0305ecfc31459c81a0e0a2d989f9dc : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54d18a8c6f2f455aaf04e1f881eacdd8 : SUCCESS in 4m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3dffd6c049684f84a270e3e91ceb5033 : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/159affe8e5574787a0ba36c0a08703b4 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4239f5e89314446a97b0c95ba23e4aa3 : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eabdfb751581433180cbe27bebb584d9 : SUCCESS in 35m 08s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6e401c2120fb499387548fb381f4ba4e : SUCCESS in 1h 03m 22s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c00df0cb0c474ee0810cb3c89e07381c : SUCCESS in 1h 41m 18s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/d6898f0f71d549ddb8f9fb3badedab78 : SUCCESS in 21m 27s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/010150483162428cafbb713d89dc1244 : FAILURE in 23m 47s","accounts_in_message":[],"_revision_number":1},{"id":"15f8c2636d76f241171c7392dc4652e579fd7190","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 11:17:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"515b73cf57160aebada442763e33a9105da046fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 12:49:35.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/6fe0b3bfc30846bb97a5607c4bdb079d : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67c4437f48e24a46b611c261583573cb : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a37ec5f0b194ca0ae25772c1f1b364e : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cfcf8151d0e446ff9bab7e4a9b47c14a : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c5b065ef815462abc76ccd6ee63a746 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9eca905c1e234d249bbbbfc120bb24e8 : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/29d1a9e5ed76427eae7bc69daa09b8c0 : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b05512fc1a8407e8d95fc26e3e9fa23 : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f2fa7abdc24c4f038245ccb88d2ea7dd : SUCCESS in 29m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e90b2a93a35548c1846b454ede944b9d : SUCCESS in 59m 35s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b48c1986dd3f48aa85a958fdeb53dbb2 : SUCCESS in 1h 31m 36s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f945d16cad844e03a2a5bea3644bab8a : SUCCESS in 23m 45s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4ab9f91f4d8045fbba4a6e50e9edff0c : FAILURE in 18m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1e8242e724acac1b27a6b181d08d4452d6944c40","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 16:06:37.000000000","message":"Patch Set 1: Workflow-1\n\nBifrost in queens uses http to download cirros, which seems to be rate limited, and times out. Fix proposed to bifrost here: https://review.opendev.org/691385","accounts_in_message":[],"_revision_number":1},{"id":"556c6ef929b56523c900da4214093cf54a6aec3c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 14:52:09.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b34e51437399ba7993c1043634f121ba5b4a826","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 14:52:13.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"56c5e3595710ae4f0653789a4e579c342940275a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 14:52:17.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a351d17ed4af8c349ede25b5e358a19f9de7ea1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 16:53:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/98cdd4242239430b8e6ab5df69413125 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7affe8744c94b60ae7df8688ad6560f : SUCCESS in 3m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aefcde720a844f4bb3b253545222e0a5 : SUCCESS in 5m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4c402fd8872c404d899716cb8f084426 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18314019c81042c88667ac2193b5e988 : SUCCESS in 4m 15s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73ce19fa667745bc81d9c674091210ae : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0760848ba07943748b33366d5c8da53b : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f4dae9634951484a9ea29d142fa044f2 : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/524b37226dc34b22ba50612ced5be7d2 : SUCCESS in 32m 24s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/deb1bfa6408e4baa825a79b8a70dcebf : SUCCESS in 1h 06m 10s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/74a35be3205744feb04fbf6259bcd365 : SUCCESS in 1h 55m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c59d08ece9834c889e2470b255e6277e : SUCCESS in 20m 16s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/35cffa2ba0244c33aa8d4b8af855ec23 : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"3a79cb62eaef24b2da1e15aa4f4f26bbfc841456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 16:53:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33ed809b51296ac8449e6c99d450b5bcd65f4f55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 19:02:42.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/be16caaee85b4aafa6c9527e40f5cfc3 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeebf7887e3f45d08e9b35412806ac46 : POST_FAILURE in 22m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9f4fb2de3cfb482d998f41a847b136b4 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b2d1f8f562a4e54b8a17b3f360ec191 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/55cd22c7f9bd4904a423b64a4ce72b91 : SUCCESS in 5m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/038abc81441e4d3b8f4d2289e11e8f8d : SUCCESS in 3m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3eddff896ef14ba1986baf4dd4cd493e : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2edd3e93783147b1bd10e91bd36c58ac : SUCCESS in 32m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d2e60947be274ed9bdf833c523c99558 : SUCCESS in 1h 00m 26s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/cf003365fbac42f99d5d5c487d3ed35f : SUCCESS in 1h 41m 59s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7344cd34e4454141b00ba1ab6ababb3f : SUCCESS in 20m 12s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bcd973f3a5d9424084f8cba3544bfd27 : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a86fbe820328b545b1e2e613d597f7340ccef255","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 19:19:36.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"57a4c5bd88d87813f54d294cbe09423310d0f0d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 21:07:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7a1e929e4af842cab05577c95cee7df1 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9046e54623954a649ac2c09d34a48d5f : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53a68b282ff147319b1376e56fe76d86 : SUCCESS in 4m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3ddfba7f29504f44afb90345acdac2b2 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecf6280f628446cc9c52c58b34ccd7ad : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7154d5fd28244da292935b6365bc4996 : SUCCESS in 4m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/71007ebc23914f99b3943c8a2d5461d3 : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb3e71ba98334bb590b584dc20ad7fa9 : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/92ff50f45185429cbfe61e62dbe29d4b : SUCCESS in 30m 34s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/903a37060a674b21988e5d181f58ad90 : SUCCESS in 1h 06m 03s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1cd8d4e553e24d7190f10206f914d9de : SUCCESS in 1h 44m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/da84fb7f3925431ca72078654a4c1f8a : SUCCESS in 21m 19s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ed5bb0bee0d0430c9181ab1546d31519 : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":1},{"id":"df95c87057647018283de494a2b074d023029c1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 21:08:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f29b9dc12f3b11ce780cc0ed3ada165b0a1c31ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 23:00:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/dfcda247820a43f2b5cb598450c8f384 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc030503ac764502b36d3fd6127bcdf0 : SUCCESS in 4m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/902a4e06245149dfbb3063ca8a435488 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55db3ddc93cb4713be8f998827f51a96 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e7283c5973c04dc4a7fc189cad5eb944 : SUCCESS in 4m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/09f672c203b04527b003f0c4413a1cbc : SUCCESS in 5m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9126ed93a10148bd98d7271935ebf3a3 : SUCCESS in 9m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d502d9fcb7a424ea4279daa2a641173 : SUCCESS in 29m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/51535fcc164e4072becd54f00363291d : SUCCESS in 1h 33m 13s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ac88d30b951346729f3941d303e95cd9 : SUCCESS in 1h 44m 13s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/360917aff66f4b0684a706e8fc550604 : SUCCESS in 19m 47s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/157dc8544e444be283312f40c810a427 : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":1},{"id":"fc09718b57a04f8eda3f9a8cda5708ce57a54876","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 23:00:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd5ab1da86d5d38e707739007e81baff85a39f7f","revisions":{"fd5ab1da86d5d38e707739007e81baff85a39f7f":{"kind":"REWORK","_number":1,"created":"2019-10-21 10:24:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/09/689709/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/09/689709/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/689709/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/689709/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/09/689709/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/09/689709/1"}}},"commit":{"parents":[{"commit":"852f7566699d9074d2f830589f9eeab78fd0fc0e","subject":"Merge \"Set provisioning and cleaning networks as non-shared\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/852f7566699d9074d2f830589f9eeab78fd0fc0e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-14 17:04:23.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-21 10:24:38.000000000","tz":0},"subject":"Try running `docker info` again if it fails","message":"Try running `docker info` again if it fails\n\nDocker sometimes fails to reply if it has just been started.\n\nChange-Id: I5ae37b8f264437a9e49b09da459e604191f9ed3e\nStory: 2006718\nTask: 37089\n(cherry picked from commit 3993d898390fdd98037e5626cdc34a6ba20a8e92)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd5ab1da86d5d38e707739007e81baff85a39f7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd5ab1da86d5d38e707739007e81baff85a39f7f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
