)]}'
{"id":"openstack%2Fkayobe~689810","triplet_id":"openstack%2Fkayobe~master~I010b2fdd7c39e3314a9e1c59072d877f635f7522","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I010b2fdd7c39e3314a9e1c59072d877f635f7522","subject":"CI: Increase job run attempts","status":"MERGED","created":"2019-10-21 14:53:48.000000000","updated":"2020-01-06 10:08:38.000000000","submitted":"2020-01-06 10:06:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"689810-1578305174590-17819b5d","meta_rev_id":"d9990e50a982eb4cf094dd82fe434e33dafea777","_number":689810,"virtual_id_number":689810,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2020-01-02 11:46:13.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 10:06:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-01-03 09:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-10-21 16:36:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-06 08:13:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-06 08:13:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-21 16:36:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-01-03 09:17:12.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-01-06 08:13:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-01-06 10:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c533b8fee75dc2c9a3497057e8475138b6c8ab6c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-21 14:53:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1b321093b1dc7c1fcf94e1ccc16bb47b118b363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-21 15:59:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1e5f2471ea249da8d9b320c5dbcb11b : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bb672acb4fea4a2b86c3bd14a611c19b : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27b7e23ea42b4ce58f3220626fbf781e : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9feece24308e49c9b6d1809fadd1ae38 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a4e37ea63931482c9f3020eec6207678 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53cda79694a3492c82dc09e93a49a469 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e11e6d0e39dc4ea7bfc9a06af46b54f8 : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e5290cf997cc4ba18a1a9737510b5c07 : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/131af3e8ac1c420a8b7f68d3d65e318d : SUCCESS in 30m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5352496e17694180a3672010e0d287c5 : FAILURE in 16m 14s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/62a1bb0358f3437ea202a686a7f1975e : FAILURE in 1h 03m 29s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1c33c52f6e9845158ccadb0adb36e170 : SUCCESS in 23m 37s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c5b98fc22d894e5da5c5f06995469b37 : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7cea9f6b38684370a0c4f3f19350c745b2975d5d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-21 16:36:07.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2eddc68bc3b78067b9733b5e4b59242efc37f916","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 12:46:23.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"84527edf632f3406875681cd40af65a307ec63f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:12:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84a33f32049b47aaa1e3bffac185e154 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/242db98d3b55469a9b852e147b1f7a38 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77f91351072c49778d0c40146abf53ee : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f42e7d17e28e4b7f8cf0068f41c4f6e2 : SUCCESS in 4m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ba465677f4a944048b375b76e04c2f6b : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53313e25381445f6958eb86c14eab53b : SUCCESS in 5m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be575748f4f04e6f920b9179ef6134b7 : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/952976f7616d4a8180b292f6dd7247a7 : SUCCESS in 12m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/42b17a7e23404b2a9a98a517b5cf30ae : SUCCESS in 37m 44s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/960856e308b8470081ef8a2c14382318 : FAILURE in 19m 49s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1fdecb4ecd4a4409951469b2ac53e5da : FAILURE in 1h 16m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5ab8a448f4ba4c29b13bf163b30c1e94 : SUCCESS in 26m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/47cf1d99dfc145eb81c23c99d61d6672 : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2f65a2732c1563e764d34640428b6f65956ab4ea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 11:16:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1ae205f11fc26538b649e9b07e58414511b3efdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 13:24:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cecb62577745406bb5b3b985e206a55c : SUCCESS in 5m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a84e441d91c4adcb93eab50eda46171 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a5a4708d1ab4ebf864d61a5942385fe : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abc9e44c1fb4b84be17357fd6a24dfb : SUCCESS in 3m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a16373ed55f946cda2076f60376de6fc : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aa07bc554c34e199a3e0658e665d738 : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af33899094a14f54a39b7afa6f4d3261 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b75cab26cbd498aa05143a30c32b98f : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c47c08a23c44e4b8ae4291b480cac2b : SUCCESS in 37m 03s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e175b56bd6fa42cb8fac9106a7411305 : SUCCESS in 1h 14m 47s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5ec76dd3a111487ba20929bcba6c23ce : SUCCESS in 2h 01m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/daaaf1a022a94d6790aa139fd2989f0d : SUCCESS in 20m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8997be41bdb74fc39f4d9d400ca1263d : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":1},{"id":"201ce9af602827bb7bbe72a3c3ec0380f9711ce2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-16 10:20:38.000000000","message":"Patch Set 1:\n\nWould this help with Galaxy timeouts?","accounts_in_message":[],"_revision_number":1},{"id":"fb2f1b1061c1ab25e51cab211fc53758c7d12952","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-12-16 10:22:03.000000000","message":"Patch Set 1:\n\nPierre - only if you do it in pre and they are not permanent. ;-)","accounts_in_message":[],"_revision_number":1},{"id":"e24ad8f4102117ba029f838a76530fe5f5b5d69e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-01-02 11:17:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48bcb62773dc792ceebfaaaeedb69208ead2d366","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-02 11:46:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c6bbb3eea6b56072e08d75b579a0f1ff1a903de7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-01-02 19:20:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54c901e83b6418c033cd5a9f71160ec414aa3284","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-01-03 09:17:12.000000000","message":"Patch Set 1: Code-Review+2\n\nOK, thanks. I guess we can move the initial deploy on the upgrade jobs to the run stage at a later date.","accounts_in_message":[],"_revision_number":1},{"id":"41be202e0a618e5b2e60346b90ba96dac2af3914","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-06 08:13:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1913aa0e88281c190e1fff4d2e7ef8d5d0522878","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 08:13:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"97f7a684a8a1cb65abb38e31154bf02ef5b087b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 10:06:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"adf2729aae5a1bb12f4d2f1bebb74a62f1116bab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 10:06:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ea0ba130cf544945986bcf3dcb4b145b : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efd68d7b6ab44c67b1fa2efc198b3ef0 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a09302f7c6443e99c91c23be056d68 : SUCCESS in 3m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9b00f4b84f748a782d6f76f2ba957be : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e29470c3d8a44a33be96cf2d7e93c223 : SUCCESS in 5m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d43de48a420a462aa1ec09c225322759 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fbf4cc1b8ce41eb8759a443e61cd3c3 : SUCCESS in 7m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/639aee97d73845659f7dea7b731943cf : SUCCESS in 36m 41s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7af7e03b33504f42a1ee81feea3a1be7 : SUCCESS in 1h 10m 47s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b543939d4fbd42eb8c311e29dea6e6bd : SUCCESS in 1h 47m 53s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1458f09f0a3d464aaea16e41702e177c : SUCCESS in 24m 03s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0517d93a14b347eab3c004467630eef2 : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":1},{"id":"d9990e50a982eb4cf094dd82fe434e33dafea777","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 10:08:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc024e3485d04947bc982e805ee1ee42 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e89518d0c1eff0439e6984f3eec07bc4f09d077","revisions":{"4e89518d0c1eff0439e6984f3eec07bc4f09d077":{"kind":"REWORK","_number":1,"created":"2019-10-21 14:53:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/10/689810/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/10/689810/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/689810/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/689810/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/10/689810/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/10/689810/1"}}},"commit":{"parents":[{"commit":"a3d3649c5ef555bfc73786258030a6986e779cc9","subject":"Install libffi headers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3d3649c5ef555bfc73786258030a6986e779cc9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-21 14:53:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-21 14:53:05.000000000","tz":60},"subject":"CI: Increase job run attempts","message":"CI: Increase job run attempts\n\nAttempts affect pre failures.  This means we can increase stability of\njobs by rejecting nodes that fail pre without failing runs at the same\ntime (unless we are really unlucky and hit b0rken nodes 5 times in a\nrow).\n\nThis change is adapted from I17b7f878c742fa8db66f738526855a02ab9f1905 in\nkolla-ansible.\n\nChange-Id: I010b2fdd7c39e3314a9e1c59072d877f635f7522\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e89518d0c1eff0439e6984f3eec07bc4f09d077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e89518d0c1eff0439e6984f3eec07bc4f09d077"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
