)]}'
{"id":"openstack%2Fkayobe~690559","triplet_id":"openstack%2Fkayobe~master~I22ab15c42d8db772f1a64f325f9371860491ccfb","project":"openstack/kayobe","branch":"master","topic":"bifrost-dib-extra","hashtags":[],"change_id":"I22ab15c42d8db772f1a64f325f9371860491ccfb","subject":"Add new variables for customising the deployment image in Bifrost","status":"MERGED","created":"2019-10-23 12:37:14.000000000","updated":"2019-12-06 18:04:21.000000000","submitted":"2019-12-06 18:04:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"690559-1575655461819-a858fa3e","meta_rev_id":"df352b8cdee7d60381fae84c03924625d1fd8db3","_number":690559,"virtual_id_number":690559,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 18:04:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-10-25 14:02:56.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-06 15:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-10-24 09:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-12-06 15:22:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-24 09:28:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-06 15:22:02.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-06 18:04:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d985c57f82678b1ac0969ee11329e4300df6489b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-23 12:37:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b1a6e48489880827baff808428d9b619370c35a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 14:16:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed81d7dcbb4b42f69ecd4a3f15796d7a : SUCCESS in 3m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/47d78d2d63ea41a9b015ec098db164e2 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e398e235e77641acb95a2aa438c102be : SUCCESS in 5m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/375e101ba56840ba9864e4a94dd6c429 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e5a92945057141c298da7ae0dfdbc66e : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b80f3f9085a456bbf58b88702bdb1ef : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c0b628019394cfd9b2487b193317092 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d04d83d6c42c4be4a0bb7c05fcf7cfbb : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7579f9c54f8543c89a951ae1c9485ba3 : SUCCESS in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/114329259ebe41f68fd5a1022d07a195 : SUCCESS in 31m 50s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/59ed51d3b1764d0398d2109cd4bd0dbb : FAILURE in 33m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2cda004888424f6d8c066c062a3a5782 : FAILURE in 1h 33m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/71cb98e86529493798e1d244c38fb886 : SUCCESS in 20m 49s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5c893a346f4945b0ae0b2e72c99d65c5 : SUCCESS in 43m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d6e39de3b0b8dc1ad05b0d00ad18973b758a8102","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-23 16:48:56.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"9dd12fe21a8cfff6d3ae2bc6fe082373a84436f8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-23 20:13:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"63d79e14b8f0979888624c36dc092fc0f8179371","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-23 20:14:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3cfa0679cb09bab56d571d3527e3aa0c3ff63019","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-23 21:47:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ee52560f8b34beeb9f8e8164683d42c : SUCCESS in 5m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a66419374407402092aaa25a1f78ece5 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6428b377241453c851a75e518611e05 : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5df17bb391394dbda5ea50f03eca5339 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c91310ec1cfe41d6ad0a8c51c96cf411 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f315ca4eb344c7871b145eb089a211 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f60949322c24ddda03f2476fa576a25 : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68beb6fe5e5e416e8ff281c4c8a8ba9b : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1af96b1b2e574d0386696704778ebf4d : SUCCESS in 7m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/21303d667f554b079e40bdb46856f081 : SUCCESS in 36m 18s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/37967bcc01e548e4bf3ff7865fcff439 : FAILURE in 17m 12s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d544265924614594b03ab026b15539fa : FAILURE in 1h 22m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3c7e8667af124897aada6ee659d2b422 : SUCCESS in 20m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d1d3ddec5603496eafbd4c3392404ca3 : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":2},{"id":"4540dd888be8ad8403153e4732ab904d566bbaaf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-24 09:28:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"48f310fdb21651dfc52ee07ee5c571b96215f6c1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-25 14:02:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"599a568c78b36a2003420cad7c154428ea32510f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 16:08:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4c7c12dc15c4e4ea530ed5579441ca5 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/882a3c7f9a1c4b5cb4df208946f6f993 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf9f0184af3a4092874d0bd8ec18165c : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a88f1d0b2d0a448f8dcbabb803898975 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/76d7220e3d0940a996d77270288d9526 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a7e7283b8347db9153f833f53b8a83 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6674b2cd117042069fcf7039e59b3563 : SUCCESS in 3m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3700ee572c2a4315b5fefc69246c6e7f : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5787c8a0a3a54ee9a3c895a181774584 : SUCCESS in 7m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6cd57d8e4c9f4d33bf6a4eae1ee402df : SUCCESS in 32m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/84be5f9cfc9a4b55ab2623d25843c8aa : SUCCESS in 1h 05m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/efc9b1f69ade4998948dd1b4feb66199 : SUCCESS in 1h 58m 47s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ab5d6df49ead4fbd85c1060ebc488b4c : SUCCESS in 21m 05s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/68b943e0dbef493389302e60e552bdf7 : SUCCESS in 44m 07s","accounts_in_message":[],"_revision_number":2},{"id":"6c776afc77634ef2f9883f9eaeea0303cb1076c5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 15:22:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"817b50f5825b6eecccd10093d05c6ca4d009c0fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 15:22:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0d90a7c18d1c56a2a0cfa34235ac6e03774ee47b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:04:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5e14ab5ef641453782e731a0172c6fff : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9b9b56c4fcb4a21ab81b8c909c198cf : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d4efc277974bf28c8dc06354ffaaf6 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8162fb6d249c4303ae85e8c8d7c73e2d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5b704273a644d9690443de93fea8984 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05860bc85fc04862b62b732822ed7aaa : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ac76f5cd09ea4720a496f726cd7402bc : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/42e76af7e0f648078199574efd664508 : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f85de38d94e0438db92f6084b3ffa54f : SUCCESS in 39m 06s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/be56658e6a204641be85a1053e623e2e : SUCCESS in 1h 18m 34s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0b385b5a2d3a45bdbcd7b1f294e463d8 : SUCCESS in 1h 51m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5feb0457763e45de9948b66240fc3dc5 : SUCCESS in 21m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a00acdf43c874b3490c2637c3c3e2107 : SUCCESS in 39m 02s","accounts_in_message":[],"_revision_number":2},{"id":"361b1a904f6faa994066bda99c5cf81499d8cedb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:04:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"461b941329df9b2264cf64acd0dc3ce1e16e9b0a","revisions":{"b42cd50ba535f5f94ba976b228dc45176f086f06":{"kind":"REWORK","_number":1,"created":"2019-10-23 12:37:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/690559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/690559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/690559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/690559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/690559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/690559/1"}}},"commit":{"parents":[{"commit":"a34065233f16e73897c23cdb214b68fb7ed16ce2","subject":"Merge \"Build pdf doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a34065233f16e73897c23cdb214b68fb7ed16ce2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-23 12:33:22.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-23 12:33:22.000000000","tz":120},"subject":"Add new variables for customising the deployment image in Bifrost","message":"Add new variables for customising the deployment image in Bifrost\n\nAdd new variables, ``kolla_bifrost_dib_elements_extra`` and\n``kolla_bifrost_dib_env_vars_extra``, in\n``${KAYOBE_CONFIG_PATH}/bifrost.yml`` for setting additional\n``diskimage-builder`` elements and environment variables for the disk\nimage generated by Bifrost. Default elements and environment variables\nare now respectively defined by ``kolla_bifrost_dib_elements_default``\nand ``kolla_bifrost_dib_env_vars_default``.\n\nChange-Id: I22ab15c42d8db772f1a64f325f9371860491ccfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b42cd50ba535f5f94ba976b228dc45176f086f06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b42cd50ba535f5f94ba976b228dc45176f086f06"}]},"branch":"refs/heads/master"},"461b941329df9b2264cf64acd0dc3ce1e16e9b0a":{"kind":"REWORK","_number":2,"created":"2019-10-23 20:13:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/59/690559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/59/690559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/690559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/690559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/59/690559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/59/690559/2"}}},"commit":{"parents":[{"commit":"a34065233f16e73897c23cdb214b68fb7ed16ce2","subject":"Merge \"Build pdf doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a34065233f16e73897c23cdb214b68fb7ed16ce2"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-23 20:13:02.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-10-23 20:13:02.000000000","tz":120},"subject":"Add new variables for customising the deployment image in Bifrost","message":"Add new variables for customising the deployment image in Bifrost\n\nAdd new variables, ``kolla_bifrost_dib_elements_extra`` and\n``kolla_bifrost_dib_env_vars_extra``, in\n``${KAYOBE_CONFIG_PATH}/bifrost.yml`` for setting additional\n``diskimage-builder`` elements and environment variables for the disk\nimage generated by Bifrost. Default elements and environment variables\nare now respectively defined by ``kolla_bifrost_dib_elements_default``\nand ``kolla_bifrost_dib_env_vars_default``.\n\nChange-Id: I22ab15c42d8db772f1a64f325f9371860491ccfb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/461b941329df9b2264cf64acd0dc3ce1e16e9b0a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/461b941329df9b2264cf64acd0dc3ce1e16e9b0a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
