)]}'
{"id":"openstack%2Fkayobe~691387","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I24a8832c02db6ba019ab8f5c2b9d7216a9b7d213","project":"openstack/kayobe","branch":"stable/stein","topic":"libffi-stable/stein","hashtags":[],"change_id":"I24a8832c02db6ba019ab8f5c2b9d7216a9b7d213","subject":"CI: Don\u0027t set cpu mode","status":"MERGED","created":"2019-10-25 16:12:36.000000000","updated":"2019-10-25 20:14:39.000000000","submitted":"2019-10-25 20:13:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691387-1572034407338-e84120e5","meta_rev_id":"d89f7f74a2def56630dfe86dc3f5fbe0387882fd","_number":691387,"virtual_id_number":691387,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-25 20:13:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-10-25 17:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-25 16:17:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-10-25 17:04:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-25 16:17:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-10-25 17:04:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-10-25 20:13:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ecaa735d01adcdd1d488c67774a6e1c96163506","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-25 16:12:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e2e1e1eb73ad7f6126b4e18e53b96c4335a7584c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-25 16:17:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45986950f8149007aa047e48c642d70929117368","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-25 16:18:07.000000000","message":"Patch Set 1:\n\nTopic doesn\u0027t seem to match patch.","accounts_in_message":[],"_revision_number":1},{"id":"be5b98dbd67c4bd4d52d871ed78d9eff6460ad0f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-25 17:04:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba364e5a24bf4f2553bcdaa259b8ff4283ecd875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 18:05:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/763f6d04ecc04c60bd96d650b5f92736 : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e313ae81cfc846fa9cf954994a15acb5 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0cdf429701d64911875b50872e0afa47 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9b59e6a3f814f69a4247e686b87b592 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3b942a0a684841848043c700747c6819 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f0881a34534aad89e41e393ce49edf : SUCCESS in 4m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3a6e5cd721a74cec9b694e6abfaf62b2 : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/166845b7ad1a48c889dae1219e46b4f3 : SUCCESS in 7m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7ad47e653c94b7f930048306cfd9b1e : SUCCESS in 30m 45s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/521609e2fe324e09a84d825c3295545b : SUCCESS in 53m 53s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5323b1c2805e4994a9af255a24191730 : SUCCESS in 1h 51m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1880541abc3c4238950e3988663662f7 : SUCCESS in 20m 35s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d91cc91b9aa146f0b34c3e2624f55dec : SUCCESS in 27m 07s","accounts_in_message":[],"_revision_number":1},{"id":"38ece0816596665804be1f4b5b279d3de2425930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 18:05:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5112623eb5370a9b2fcf412a0135a296517f79e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:13:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/320f1bcfeb7b4dbfa375aaa010172f14 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84a6ebe6b2264137be30fc23caa1f849 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e12b1ca434e9456d9846620bd141e91c : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d93767472a6d4cd696164d3c37610b6e : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1599ee81d6a845949d91c77ba57cc8b5 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19fb333212b0403aafe99803e6e456c1 : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/beb227def40c4c49b29edac80e780c14 : SUCCESS in 8m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f08c71aab5140969fbcf932076dc392 : SUCCESS in 48m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/030f2bfe11ce43abbee5efc0717391bf : SUCCESS in 1h 07m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8ca1446fc8754b97861e7d753433cf4c : SUCCESS in 2h 02m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9b0aa746f6404313ab17d40a53ecc8a7 : SUCCESS in 27m 08s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f2f45adcbad9418bae9e3dbe52c811fa : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"2902ed4c8c7dfcb55d5035a51deadbe5fe348209","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:13:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dd71a98faca2321c9889ec80b7a19b43020278fa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 20:14:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27a9b8499544417b4d6584e81a58d3e : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2d280086d511bc63d03a774761351f2ae71c1266","revisions":{"2d280086d511bc63d03a774761351f2ae71c1266":{"kind":"REWORK","_number":1,"created":"2019-10-25 16:12:36.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/87/691387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/691387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/691387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/691387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/691387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/691387/1"}}},"commit":{"parents":[{"commit":"91af67de4c6ff04bc0d12fd24d86790fd4098a58","subject":"Try running `docker info` again if it fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/91af67de4c6ff04bc0d12fd24d86790fd4098a58"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-16 13:52:57.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-25 16:12:36.000000000","tz":0},"subject":"CI: Don\u0027t set cpu mode","message":"CI: Don\u0027t set cpu mode\n\nNot setting cpu-mode will reduce the number of TCG emulated features.\nHopefully this will make CI more reliable.\n\nChange-Id: I24a8832c02db6ba019ab8f5c2b9d7216a9b7d213\n(cherry picked from commit cc71fd03f9cd7096361efecc8e1eed6b0862eed2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d280086d511bc63d03a774761351f2ae71c1266"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d280086d511bc63d03a774761351f2ae71c1266"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
