)]}'
{"id":"openstack%2Fkayobe~691405","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I39b59e3fed94d4fc56514c98ae1d9d1c0ad6bebc","project":"openstack/kayobe","branch":"stable/rocky","topic":"underscore-stable/rocky","hashtags":[],"change_id":"I39b59e3fed94d4fc56514c98ae1d9d1c0ad6bebc","subject":"Example should use hyphen instead of underscore","status":"MERGED","created":"2019-10-25 17:43:37.000000000","updated":"2019-11-01 18:24:15.000000000","submitted":"2019-11-01 18:24:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691405-1572632655371-2bb4bfd8","meta_rev_id":"3ef627702ec9a7462d3bf228498806f3d77a609a","_number":691405,"virtual_id_number":691405,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-01 18:24:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-11-01 16:46:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-11-01 16:46:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-01 16:46:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-01 18:24:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"280aff0b06fbd052eff38bb33ff2a516d1e0ee1f","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-10-25 17:43:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b875332e64f41b2629ff3f98f063efe307bf431a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 19:21:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ce264da57b2749fda827eb48a8c30cd5 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d2f7b5063374861952171cd664eb49a : SUCCESS in 3m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa7e372b12334c9fbaf9c917804adf17 : SUCCESS in 4m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3d49d3f70e714f4bb2e7427bca21d8ca : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12dd2193c3fe4d02b9f53a44388ee029 : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c157abae1b044fc88fdec20a80c15e02 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/443e9258750b4884b0a43488d0fa8cdb : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d5690dbd87c24077ac4e800b3d3ba5f9 : SUCCESS in 7m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8fb61ea62a0341bca1ed01ee2e5fc4fb : SUCCESS in 26m 12s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/60f5648d57904a299a35b2438960c2d4 : SUCCESS in 56m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/838d23282eed41f6b7c042459a15c88d : SUCCESS in 1h 36m 43s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4358f3014e574597999ac38b6d450dbc : SUCCESS in 19m 33s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7c9766e4e260479a96eb042b00fea765 : FAILURE in 18m 17s","accounts_in_message":[],"_revision_number":1},{"id":"03a9028af99af6c0e6f70d2074771d2d0221ce0d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 14:52:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"025947450f7171fd1da324d66fc53f2475917e4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 16:43:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c80aa015f9b346b2b9f9308f599a8df5 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a8937560a14f4f9c20bf999e06ed00 : SUCCESS in 5m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0050594d26a8432a9307f412f9f38074 : SUCCESS in 5m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9ed41d1b46df4b59b8daffc7089a80ec : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e9174a91374087809e6b3db27d0e8e : SUCCESS in 5m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3a62191b24db4926873e78107f1cdedf : SUCCESS in 5m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/14a771957e8d4aacbc6241f1a9018141 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc1fabbb25a54afe9c2883ff19028ae8 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8bfd06735f634dccaeacfd9ba47b532c : SUCCESS in 29m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5a64eaae3a564327bd251bf5c7c40fe7 : SUCCESS in 1h 06m 48s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c3c9205416b44961a57537e81396045a : SUCCESS in 1h 41m 20s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f16e2ebd79b6416c8d40a439d190ce1e : SUCCESS in 19m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/69bce60947b345b88b7774841443d4d9 : SUCCESS in 33m 27s","accounts_in_message":[],"_revision_number":1},{"id":"d8b58f3af8b02cb2c07e58624f69a6b065cce315","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 16:46:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dd7de791d9c36fb44382c6f05696cccda3d86637","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 16:46:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37a6bcfe348d1ca01354017cc330e2ab18b9b1c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 18:24:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/d78e7d0dbea64b12b92685880f91d815 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/434585febd0d431ca0325c4489272d28 : SUCCESS in 3m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7ba07982d718408dba1ab30be8449a3b : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875d161ddc0d4925b83facaef01a9cf6 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9c16fef25e354a1bacf36abec8e18f2b : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1a9c3e3631d54f5e98ab8dc32bde2610 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee26e2fc7c7348e3b9ffd36135738fb8 : SUCCESS in 7m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6be27ad3d39a444f80565327e4355629 : SUCCESS in 23m 53s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/10817cb6d9c2467f9378a54e0b9902c2 : SUCCESS in 1h 07m 53s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/98efed9602534198a097841069e39c49 : SUCCESS in 1h 36m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/434ef56f9d7944f280ea5ea411cff972 : SUCCESS in 17m 43s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0fdbd1c9773047cd9da50629863c18b1 : SUCCESS in 29m 26s","accounts_in_message":[],"_revision_number":1},{"id":"f6566d6024743e44d9497cc4525e7aff71aaec19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 18:24:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a2f8e8bb6abf3ad9b27cf35180457d2b8d2dc0fd","revisions":{"a2f8e8bb6abf3ad9b27cf35180457d2b8d2dc0fd":{"kind":"REWORK","_number":1,"created":"2019-10-25 17:43:37.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/05/691405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/691405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/691405/1"}}},"commit":{"parents":[{"commit":"852f7566699d9074d2f830589f9eeab78fd0fc0e","subject":"Merge \"Set provisioning and cleaning networks as non-shared\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/852f7566699d9074d2f830589f9eeab78fd0fc0e"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-10-24 12:34:07.000000000","tz":60},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2019-10-25 17:43:37.000000000","tz":0},"subject":"Example should use hyphen instead of underscore","message":"Example should use hyphen instead of underscore\n\nAs per the kolla_sources documentation:\nhttps://docs.openstack.org/kayobe/latest/configuration/kolla.html#source-code-locations\n\nChange-Id: I39b59e3fed94d4fc56514c98ae1d9d1c0ad6bebc\n(cherry picked from commit 9c9e45600664664737b2e3790bef92373aedd278)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2f8e8bb6abf3ad9b27cf35180457d2b8d2dc0fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2f8e8bb6abf3ad9b27cf35180457d2b8d2dc0fd"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
