)]}'
{"id":"openstack%2Fkayobe~691406","triplet_id":"openstack%2Fkayobe~stable%2Fqueens~I39b59e3fed94d4fc56514c98ae1d9d1c0ad6bebc","project":"openstack/kayobe","branch":"stable/queens","topic":"underscore-stable/queens","hashtags":[],"change_id":"I39b59e3fed94d4fc56514c98ae1d9d1c0ad6bebc","subject":"Example should use hyphen instead of underscore","status":"MERGED","created":"2019-10-25 17:43:56.000000000","updated":"2019-11-06 08:34:20.000000000","submitted":"2019-11-06 08:34:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691406-1573029260875-9a54f4b4","meta_rev_id":"ebb5046d3ba063ceba60c53986954d10d0e022c0","_number":691406,"virtual_id_number":691406,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-06 08:34:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-11-05 16:05:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-06 06:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-06 06:59:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 16:05:11.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-06 06:59:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-06 08:34:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e7eb94acddb18d2395e640a08e180d7d87790d9","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-10-25 17:43:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"950fe16ca6b971af7ea0d2c8429eb81ad15557a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-25 19:08:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/281906a73a4f489290caea89f63482db : SUCCESS in 4m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68ed6c64ce7041a581d90191ce2300b3 : SUCCESS in 4m 24s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/84a8994fbe28487da1147571686d8e91 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9de4010924d9407da0522856f311089c : SUCCESS in 4m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd162a8ed6a74f9abf04e90cefdadf82 : SUCCESS in 5m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/19481adc2669440282061b3d9d7b966f : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03d22b1253d84f4eb345ffacb6b9f74e : SUCCESS in 7m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4f7c8e4a1ab04beaae3abe98b5934f3b : SUCCESS in 24m 53s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b064336b86f542b0a73aaff39ad84599 : SUCCESS in 51m 10s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1c2d847cf5bf4c82bcf7586c14e554b3 : SUCCESS in 1h 23m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c4dea10b5c754d4aa88db30e52b554ba : FAILURE in 19m 11s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/706d3871c51f43fe801bd68f917cabd5 : FAILURE in 18m 29s","accounts_in_message":[],"_revision_number":1},{"id":"af32084d1a742d1a363c30ffc74d3d4b9c980ca6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 16:05:11.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0401e40758f56e998676166485e69a270089519f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 17:02:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e60d739df30547019802d5f8b29c35a6 : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46d9b3b8098942d2b350b26a85bae6b4 : SUCCESS in 4m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a7a0c33c87704d4aa3a1a36081c9a7d0 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907524cb95d5458dbe4286f7805fbd9e : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0d69c3e2e6e498a808c59de08050287 : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a7616afceba6484fbae371b8a92ef3c6 : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e40e2f8516814d43a617951e56b90899 : SUCCESS in 8m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f463b6a7635743a38b8ef75b10a49257 : SUCCESS in 26m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/12a126e960364d38a5c679c73d1da426 : SUCCESS in 53m 01s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e858c6b2060f49dbbfd2f0cb7ca07c54 : FAILURE in 50m 32s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/227ad1f306b24d2a8399b6b0401b493f : SUCCESS in 23m 58s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/6d2af743726040638be63fe59863b221 : SUCCESS in 37m 50s","accounts_in_message":[],"_revision_number":1},{"id":"ad2b6b546bc4d176c72a56edcd3445ca6f6d9825","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 17:05:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"95d10f256b2f78d6b8778609938fff974250450c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 18:28:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/60dadcd93c4c4f828d05d97a4074fae8 : SUCCESS in 3m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fd736142cf346099c10a74896778553 : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d55394adac574894b0672fe2308fe366 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777a1ce098ab42a8ae35d45fe0fee3d3 : SUCCESS in 3m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d6a1e1e3f714513b9b481fa5d728a8c : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/617e88094e3c440db6e82ef08e66ddc9 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/716f1f94091046598c36c3ae2e9184ec : SUCCESS in 8m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c9966486968f45a78aa2aec0f144c10f : SUCCESS in 29m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/484fbe0524e44fe3ad5ed6a634a05f77 : SUCCESS in 48m 21s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8e11273506d44fd09c22a06026cf86e3 : SUCCESS in 1h 19m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/a5bf7b4969604604a168d1e226e06ed6 : SUCCESS in 24m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/42fe37ac233c4aa29df6f60ee88a219d : SUCCESS in 40m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1874e1ca2ddc6b97e7ab289444a3810a90e7ec98","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-06 06:59:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71493348dbdf3bb3e7fb287b4d229c11425f1fec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 06:59:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5bdc148baed6b7ca14ae846fb4c7aa957ae19727","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 08:34:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cc9e5ac59174a5cd48eb4a98f865d52940112eb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-06 08:34:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/08d82eac0da346699e2c45e06937880c : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a78320deb3894ee1b296def2639828c7 : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6699df76c2b74251bd4f47809832695b : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3263e23506f444d99d0bad25a91ec048 : SUCCESS in 9m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5556ea01ed3b401ab06f47f4018d53f8 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/549a5bdf402646f7b39cf7c9e26c0120 : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f605a24aa6ec400bb9a6c6be978f7d84 : SUCCESS in 30m 53s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bbdf80dcaf2e4e77a86bb95f1ff23a4a : SUCCESS in 47m 05s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/17e3b8fb84db470eb669edf4ebafa459 : SUCCESS in 1h 29m 43s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/a11836eb7ae446d6a5eeb4a6ead85e08 : SUCCESS in 21m 03s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d71c1f8594694313b14381c8a1e827b3 : SUCCESS in 43m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a9a34fffddc9255037ab5f2726a186b03fb6e5cb","revisions":{"a9a34fffddc9255037ab5f2726a186b03fb6e5cb":{"kind":"REWORK","_number":1,"created":"2019-10-25 17:43:56.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/06/691406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/06/691406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/691406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/691406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/06/691406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/06/691406/1"}}},"commit":{"parents":[{"commit":"a752c10fb8e169802ef9ae94d441092999572559","subject":"Try running `docker info` again if it fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a752c10fb8e169802ef9ae94d441092999572559"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2019-10-24 12:34:07.000000000","tz":60},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2019-10-25 17:43:56.000000000","tz":0},"subject":"Example should use hyphen instead of underscore","message":"Example should use hyphen instead of underscore\n\nAs per the kolla_sources documentation:\nhttps://docs.openstack.org/kayobe/latest/configuration/kolla.html#source-code-locations\n\nChange-Id: I39b59e3fed94d4fc56514c98ae1d9d1c0ad6bebc\n(cherry picked from commit 9c9e45600664664737b2e3790bef92373aedd278)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9a34fffddc9255037ab5f2726a186b03fb6e5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9a34fffddc9255037ab5f2726a186b03fb6e5cb"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
