)]}'
{"id":"openstack%2Fkayobe~691505","triplet_id":"openstack%2Fkayobe~master~Ib3133820a3d3ca3027ea2f9bc9b6907929cff8ed","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ib3133820a3d3ca3027ea2f9bc9b6907929cff8ed","subject":"tox: Keeping going with docs","status":"MERGED","created":"2019-10-27 06:39:25.000000000","updated":"2019-11-28 17:22:23.000000000","submitted":"2019-11-28 17:20:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"691505-1574961642443-d86a0cf6","meta_rev_id":"955dc450d6eac0d1ac3182c7008819ae11c9674b","_number":691505,"virtual_id_number":691505,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-28 17:20:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-11-11 14:52:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-11-13 07:01:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-28 16:37:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-28 16:37:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-27 08:12:12.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-11-11 14:52:02.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-11-13 07:01:39.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-11-28 16:37:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-28 17:20:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78d5f3d49f9ce0c601b4133b2ebd47a7ec3f2a86","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-10-27 06:39:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24a1cdac581ed9be60657da1771583b4d111886b","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-10-27 06:42:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"acd164037cf848b443ab88e068baf7567a70870b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-27 07:24:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96d7c942de124252a4f9a6761ccc8f09 : SUCCESS in 4m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3814a8a19f114ada9347d08beef3738a : SUCCESS in 3m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5589318740e4018a19fb834b38a6e93 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54b02c3c4fb8446ebf818015f04efb87 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/61ad25260d3548e28f948d97b3bcba8b : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d234989bfc5d4c91986b70ea5498f4ca : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cf38f0cd8f2423eaa555acce4457be0 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a6e1a067e990489baa4eff08f97283c6 : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0c3d8714892049af8c3ce847420a32c2 : SUCCESS in 7m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a0730c5a7cac4c569d76fad1c9855291 : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":2},{"id":"ab18f762726c0c515db65bcdcd72eb920b9a6b6f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-27 08:12:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5e7e42f8e8498aa92510a1b104fdb84ce69f79fa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 11:13:07.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f80e08e6fc4313a427072327206f94fb70a9efbb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-04 10:13:53.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"67c1e11bb7fbe3dc27e29fd0343e44811fcc95b8","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-11-11 13:48:08.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"e7cabf9fc8a0f45673b43c4e8bf894d467f521eb","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-11-11 13:49:58.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"025db3dbe73f128ab0bdc19d77ad6db76d8b6ddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 14:31:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35dc7a8657e94ba8904a2de0c675e898 : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/36ef58a28f6640aaa11412283f4db4fa : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/654e9c5a659f46cd8f4a958c7c4a0d12 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4224fac525d42b697196b8452247b23 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fc5148a9ad1345ec9ca8cd327dfdbc38 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/766cc6a2452b40bf91bdf50951d1b72a : SUCCESS in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84aacc9c5a664df8a87caa42aa761473 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8104b7c6bbe244638098b249b328956d : SUCCESS in 4m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/474baf2ca927480089e83ec0f8aee63b : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/40534244eb8949eeb4f7e4159aa88a4f : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":4},{"id":"ee4b1df912fdcbe14ffd7bfea0c331788df67a99","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-11 14:52:02.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":4},{"id":"f1656d70d0d8b07b4c933213f4a32c748a5c8197","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-11-13 07:01:39.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"7491aa390b22e73f6987cd38b2431767b0a0df09","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-13 14:10:34.000000000","message":"Patch Set 4:\n\nI would like to see what is the outcome of the discussion at http://lists.openstack.org/pipermail/openstack-discuss/2019-November/010711.html","accounts_in_message":[],"_revision_number":4},{"id":"63aef59fe934effa032b4ee0bae7f345393ce7ef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-28 16:37:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThe mailing list discussion didn\u0027t go further. Let\u0027s merge.","accounts_in_message":[],"_revision_number":4},{"id":"60f4bfb29e23ae3f841b7b5253cd8a830e40ada7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 16:37:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0ade234e65c9e03271a90197aac8cdadf023099f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:20:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/851ad1a700b14e938910ad048c3f4576 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06701d63359645b183bcd0bbf017c867 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17e0618f61fa4a57afb8c0d1ec3bda44 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ab0053c0c68469792b81eeda99117a2 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6742a956bafe4f92b7704e32e3dce428 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e64f6d03859e4f93a97ed0cd08cc0965 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/88d897b0bf434771a49b1d31f2f393b1 : SUCCESS in 8m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc65f7dc28044d12bb2b474e9e207997 : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01230816dd324828b5f16e46bc6a2ce4 : SUCCESS in 39m 42s","accounts_in_message":[],"_revision_number":4},{"id":"2a80a70914d43e824e31c87fa3c155bf05000a7c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:20:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"955dc450d6eac0d1ac3182c7008819ae11c9674b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:22:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80fa845e7242447185076d0539163812 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d055306d503f4c619a747bb258cb414e : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d23ed9258814b2321e758ace53c62c90fadcda0f","revisions":{"ee670138ca8fe888252ad925d55481869fa839d5":{"kind":"REWORK","_number":1,"created":"2019-10-27 06:39:25.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/05/691505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/691505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/691505/1"}}},"commit":{"parents":[{"commit":"3c98bfa5b2ec82d43ace25575b39777a81659654","subject":"Merge \"Add ansible tag headers to Host Configuration doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c98bfa5b2ec82d43ace25575b39777a81659654"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-10-27 06:39:09.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-10-27 06:39:09.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nChange-Id: Ib3133820a3d3ca3027ea2f9bc9b6907929cff8ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee670138ca8fe888252ad925d55481869fa839d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee670138ca8fe888252ad925d55481869fa839d5"}]},"branch":"refs/heads/master"},"96d5a20c10b4cebb5dceaad2c9f8355cfa7f2039":{"kind":"REWORK","_number":2,"created":"2019-10-27 06:42:41.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/05/691505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/691505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/691505/2"}}},"commit":{"parents":[{"commit":"3c98bfa5b2ec82d43ace25575b39777a81659654","subject":"Merge \"Add ansible tag headers to Host Configuration doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c98bfa5b2ec82d43ace25575b39777a81659654"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-10-27 06:39:09.000000000","tz":480},"committer":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-10-27 06:42:19.000000000","tz":480},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nSphinx 1.8 introduced [1] the \u0027--keep-going\u0027 argument which, as its name\nsuggests, keeps the build running when it encounters non-fatal errors.\nThis is exceptionally useful in avoiding a continuous edit-build loop\nwhen undertaking large doc reworks where multiple errors may be\nintroduced.\n\n[1] https://github.com/sphinx-doc/sphinx/commit/e3483e9b045\n\nChange-Id: Ib3133820a3d3ca3027ea2f9bc9b6907929cff8ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96d5a20c10b4cebb5dceaad2c9f8355cfa7f2039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96d5a20c10b4cebb5dceaad2c9f8355cfa7f2039"}]},"branch":"refs/heads/master"},"18e3dbcb91cbc0436beecb21f2c6a30d74b143cd":{"kind":"REWORK","_number":3,"created":"2019-11-11 13:48:08.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/05/691505/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/691505/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/691505/3"}}},"commit":{"parents":[{"commit":"3c98bfa5b2ec82d43ace25575b39777a81659654","subject":"Merge \"Add ansible tag headers to Host Configuration doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c98bfa5b2ec82d43ace25575b39777a81659654"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-10-27 06:39:09.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2019-11-11 13:47:50.000000000","tz":0},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\nSphinx 1.8 introduced [1] the \u0027--keep-going\u0027 argument which, as its name\nsuggests, keeps the build running when it encounters non-fatal errors.\nThis is exceptionally useful in avoiding a continuous edit-build loop\nwhen undertaking large doc reworks where multiple errors may be\nintroduced.\n\n[1] https://github.com/sphinx-doc/sphinx/commit/e3483e9b045\n\nChange-Id: Ib3133820a3d3ca3027ea2f9bc9b6907929cff8ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18e3dbcb91cbc0436beecb21f2c6a30d74b143cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18e3dbcb91cbc0436beecb21f2c6a30d74b143cd"}]},"branch":"refs/heads/master"},"d23ed9258814b2321e758ace53c62c90fadcda0f":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2019-11-11 13:49:58.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/05/691505/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/05/691505/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/05/691505/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/05/691505/4"}}},"commit":{"parents":[{"commit":"3c98bfa5b2ec82d43ace25575b39777a81659654","subject":"Merge \"Add ansible tag headers to Host Configuration doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c98bfa5b2ec82d43ace25575b39777a81659654"}]}],"author":{"name":"gujin","email":"gu.jin@99cloud.net","date":"2019-10-27 06:39:09.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2019-11-11 13:49:54.000000000","tz":0},"subject":"tox: Keeping going with docs","message":"tox: Keeping going with docs\n\n1. Sphinx 1.8 introduced [1] the \u0027--keep-going\u0027 argument which, as its name\nsuggests, keeps the build running when it encounters non-fatal errors.\nThis is exceptionally useful in avoiding a continuous edit-build loop\nwhen undertaking large doc reworks where multiple errors may be\nintroduced.\n\n2. Cleanup the folders before run commands\n\n[1] https://github.com/sphinx-doc/sphinx/commit/e3483e9b045\n\nChange-Id: Ib3133820a3d3ca3027ea2f9bc9b6907929cff8ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d23ed9258814b2321e758ace53c62c90fadcda0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d23ed9258814b2321e758ace53c62c90fadcda0f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
