)]}'
{"id":"openstack%2Fkayobe~691590","triplet_id":"openstack%2Fkayobe~master~I0cb7af6d4bae1a3e262f70304c301ec459871e9b","project":"openstack/kayobe","branch":"master","topic":"drop-py2","hashtags":[],"change_id":"I0cb7af6d4bae1a3e262f70304c301ec459871e9b","subject":"Drop python2.7 support","status":"ABANDONED","created":"2019-10-28 06:01:19.000000000","updated":"2019-11-28 03:31:13.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a6e1b10190a2404e300b855582f1cfc1c06d3923","_number":691590,"virtual_id_number":691590,"owner":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-28 11:18:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-28 11:18:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6ffbe5ead9b10631b27d9ebc80678d673142e64b","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-10-28 06:01:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d673428dadb5bcbd18d2150964119e56684ce62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 07:57:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/790662e07a7246cab825ff762ba936de : SUCCESS in 3m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22ccefc22ca54248a2f699eebed3a263 : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/82e162c7380b4a4eb41325d2c9eed2d2 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e19f891cfd68432dbc4fe964ab384fb5 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13792b4d7c794615a809f5ef7d591165 : SUCCESS in 7m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e6551289a294905883b48a295204661 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5f903fc09ebb4f66a089b6f1fd23e6ba : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eac3b3929d444ed2b9628c253e5dab57 : FAILURE in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/42d11233d5114c108b7edaa18735a0fd : SUCCESS in 39m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/00491c1636e745d89a5921db813b9d41 : SUCCESS in 1h 10m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1e1118193d684f0db02383c282a985ac : SUCCESS in 1h 50m 28s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/117c1e68dfc141779d225c85448092c8 : SUCCESS in 20m 37s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7e74e76561964a12915ddc5239a59e59 : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":1},{"id":"4fbf3003c70d712521e22401cc95f0b4c181ac88","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-10-28 08:02:36.000000000","message":"Topic set to drop-py2","accounts_in_message":[],"_revision_number":1},{"id":"adc3f1245e55fef314ecd2636a1254f6e96f7b62","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-10-28 09:15:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7df98fcf7c5d3128ea1ac38fca6125fa85e642e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-28 11:18:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5675f7eb7d7e41eca449ae31f4ed54cb : SUCCESS in 2m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f2bd49801824017bbad53bdbea4eec3 : SUCCESS in 4m 56s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f62697ecfe5e40adb654a9ab260ada72 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/da8b745327df45ba995047f22055fdc7 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49ad430380a94d609ad0ed809867187b : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9b5dbb299d3e4a9dae1ba9b1bb4ca4c2 : SUCCESS in 3m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0e075bcf3bc14f35ab1f4dc4a535068c : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b1734807099c44c8a48c719600c02b4f : FAILURE in 9m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08dbe244778743d685df53d790143b4d : SUCCESS in 37m 49s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/67f52a77a855498facba935170cc17b0 : SUCCESS in 1h 11m 42s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/57fb0b27c6f1429bbe4920e3d6b130cd : SUCCESS in 1h 57m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1008ecc960384db78a57c2a3ad608304 : SUCCESS in 21m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d5b8a57bf39c47868c7b4f3ba2f405f2 : FAILURE in 27m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a6e1b10190a2404e300b855582f1cfc1c06d3923","author":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"date":"2019-11-28 03:31:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fd36b18ccdd4c86de8afab4746b005236b171606","revisions":{"c4eb400eafe9032fa29b27a1284bb7717ec99fad":{"kind":"REWORK","_number":1,"created":"2019-10-28 06:01:19.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/90/691590/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/691590/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/691590/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/691590/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/691590/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/691590/1"}}},"commit":{"parents":[{"commit":"3c98bfa5b2ec82d43ace25575b39777a81659654","subject":"Merge \"Add ansible tag headers to Host Configuration doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c98bfa5b2ec82d43ace25575b39777a81659654"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-28 05:58:35.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-28 05:59:58.000000000","tz":480},"subject":"Drop python2.7 support","message":"Drop python2.7 support\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid\noff.\n\nChange-Id: I0cb7af6d4bae1a3e262f70304c301ec459871e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4eb400eafe9032fa29b27a1284bb7717ec99fad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c4eb400eafe9032fa29b27a1284bb7717ec99fad"}]},"branch":"refs/heads/master"},"fd36b18ccdd4c86de8afab4746b005236b171606":{"kind":"REWORK","_number":2,"created":"2019-10-28 09:15:02.000000000","uploader":{"_account_id":30407,"name":"haixin","email":"haixin_haixin@qq.com","username":"haixin"},"ref":"refs/changes/90/691590/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/90/691590/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/691590/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/691590/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/90/691590/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/90/691590/2"}}},"commit":{"parents":[{"commit":"3c98bfa5b2ec82d43ace25575b39777a81659654","subject":"Merge \"Add ansible tag headers to Host Configuration doc page\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c98bfa5b2ec82d43ace25575b39777a81659654"}]}],"author":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-28 05:58:35.000000000","tz":480},"committer":{"name":"haixin","email":"haixin@inspur.com","date":"2019-10-28 09:13:18.000000000","tz":480},"subject":"Drop python2.7 support","message":"Drop python2.7 support\n\nTrain was the last release supporting python2. Let\u0027s rip the bandaid\noff.\n\nChange-Id: I0cb7af6d4bae1a3e262f70304c301ec459871e9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd36b18ccdd4c86de8afab4746b005236b171606"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fd36b18ccdd4c86de8afab4746b005236b171606"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
