)]}'
{"id":"openstack%2Fkayobe~692335","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ice26a113f9023b1a25a4af0552d07c5e95e64556","project":"openstack/kayobe","branch":"stable/stein","topic":"story/2006662-stable/stein","hashtags":[],"change_id":"Ice26a113f9023b1a25a4af0552d07c5e95e64556","subject":"Fix baremetal serial console autoenable","status":"MERGED","created":"2019-10-31 11:13:49.000000000","updated":"2019-11-04 16:36:23.000000000","submitted":"2019-11-04 16:35:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692335-1572885313591-8bdcbf74","meta_rev_id":"39e3729633b4249c2435a13087c4b52804c666ae","_number":692335,"virtual_id_number":692335,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-04 16:35:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-31 13:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-11-04 14:29:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-31 13:05:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-04 16:35:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4c20fd4381a436a1cbd9374dc76a8fd4e5271d8d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 11:13:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cf9cabaf46a4163cd6d56f6f78dd0fc742f099d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 12:57:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8623946a23324296bfed02d4b784e021 : SUCCESS in 4m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/999f108cd32344d6b8fd4088d86a474c : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0398ca32f7324188a7287ec77317387c : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee9054d0bf64be5aa9c5b43c328e483 : SUCCESS in 3m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/482ecd134b48454093bcaf49b21efd07 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22c1719f546d4851a792f56521a96bfa : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21e8b6196bc84879879e1abba3561ba5 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cbfe8d0aa6294b38a962269b961b3ae5 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/be973481e2a349e3ae12d3802ba95129 : SUCCESS in 7m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee9eeeb18c69471b87fe970e84ed3873 : SUCCESS in 30m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ebc4e7d59ba247a994dd3ce74b4979be : SUCCESS in 56m 19s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d62054a1324443c9b0eb830d2f73e9bf : SUCCESS in 1h 40m 43s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0992df83b148497a88c9d83ab9b28bcb : SUCCESS in 22m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/faeea6aeb77745fbb2b87a1f6a4c4d2f : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":1},{"id":"96084cd5dea2ba49b175ebefd1ee321efad2e4cf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-31 13:05:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"29aa8021d7151e57d0c41064e1a20ec1f090d7a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:29:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b06a4fa10ddb3f871cab18071abcec34b095f1ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 14:30:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fee94751b808e6fe55d3529df5e8e496c4703de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:35:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/adcf7233444442e7913433c6d7f620f6 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5226546359d14b65b906029f1b9f5168 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a905c22a5ee4a7da66e7dbf00da400d : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ff801175db0e4a34831e611b6612adb2 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95c4360a7f60459e843cfedadcdd6317 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5c94da5ab414ba08979c768a2eff186 : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4ef02b2641d43dc9e7bd454d5badb69 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/779bed0e108d4ef28d0cf51c538f882e : SUCCESS in 8m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2df131066a224cbda30222253089910e : SUCCESS in 38m 17s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/42aeb01cf05b42d2ae36f4e66e57239e : SUCCESS in 57m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d65bfe77f61446a19a150945737f532a : SUCCESS in 1h 52m 53s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b5c545f8594b4d15980a46a69697ef2e : SUCCESS in 27m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c7422deeb964405ab394db191727ff76 : SUCCESS in 32m 21s","accounts_in_message":[],"_revision_number":1},{"id":"bf647bf989275f16026d65f69461c9c40c53d902","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:35:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"66b8730762c09c9b6135be2f885b6fa7f78a730e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:36:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c19a3e5127748a2af0d47466d654e67 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e85681d870204f7eb8c4badfd27b0fbf : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1451f67df23934e88190f178027903a2474815dd","revisions":{"1451f67df23934e88190f178027903a2474815dd":{"kind":"REWORK","_number":1,"created":"2019-10-31 11:13:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/35/692335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/692335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/692335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/692335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/692335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/692335/1"}}},"commit":{"parents":[{"commit":"2d280086d511bc63d03a774761351f2ae71c1266","subject":"CI: Don\u0027t set cpu mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d280086d511bc63d03a774761351f2ae71c1266"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-04 15:49:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-31 11:13:49.000000000","tz":0},"subject":"Fix baremetal serial console autoenable","message":"Fix baremetal serial console autoenable\n\nIf ironic_serial_console_autoenable is set to true, ironic compute node\nserial consoles are automatically enabled during kayobe overcloud post\nconfigure. However, the baremetal-compute-serial-console.yml playbook\nexpects a variable called cmd to be defined, and set to \u0027enable\u0027 in\norder to enable serial consoles. This patch fixes the issue.\n\nWe also add a conditional to only import the playbook when ironic is\nenabled in the overcloud.\n\nChange-Id: Ice26a113f9023b1a25a4af0552d07c5e95e64556\nStory: 2006662\nTask: 36924\n(cherry picked from commit 5977a4411921c1880ee023b945ac2cceb533749a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1451f67df23934e88190f178027903a2474815dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1451f67df23934e88190f178027903a2474815dd"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
