)]}'
{"id":"openstack%2Fkayobe~692336","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Ice26a113f9023b1a25a4af0552d07c5e95e64556","project":"openstack/kayobe","branch":"stable/rocky","topic":"story/2006662-stable/rocky","hashtags":[],"change_id":"Ice26a113f9023b1a25a4af0552d07c5e95e64556","subject":"Fix baremetal serial console autoenable","status":"MERGED","created":"2019-10-31 11:14:07.000000000","updated":"2019-11-01 18:15:02.000000000","submitted":"2019-11-01 18:15:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692336-1572632102554-4445955c","meta_rev_id":"d967a583baaebb54141c27936b49104d3feb7fc6","_number":692336,"virtual_id_number":692336,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-01 18:15:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-11-01 16:40:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-31 13:08:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-11-01 16:40:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-31 13:08:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-01 18:15:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"59aac3f2b0456906c30b2b0a1b664c4aac6acb01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-31 11:14:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b46ccba3f2b02ea5e5dd435869d488534d769ad4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-10-31 13:08:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dba7742d335767344bcb28278a66c1484d805f74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-31 13:14:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2b017a261e5b4028859a422f9723e1e5 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852e1f5e91954a18adcc76ea431e485d : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b55e3c39d20f4441898268b08e2d9c63 : SUCCESS in 4m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a3ff9d6900a842fba7216120b6a37e1b : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c815d0fbc8e0493b9fff3281fca5085b : SUCCESS in 3m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e63fe9b5cafd4fc7bf3cbf40ffdc5c8f : SUCCESS in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cca79648530e41feb6c665524ca18dce : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/28e715e8f88c4bf086587cdc3f34a26c : SUCCESS in 3m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f38916df1452400a94cfa51334b1945b : SUCCESS in 6m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e1d13452eb4e497bafe73932fc4f8eca : SUCCESS in 27m 45s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9ecf5938b0744cd0bac3e14874f18230 : SUCCESS in 1h 00m 53s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c6dd9341933448f18c1655b355c98417 : SUCCESS in 1h 49m 08s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/490fb96c97f040e9b62db370968e2b02 : SUCCESS in 22m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ff1f1191af7c4b26b5252142bb34d4e2 : FAILURE in 13m 13s","accounts_in_message":[],"_revision_number":1},{"id":"77c6d95beee557930f22fa9314f5c3e442901207","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 14:51:46.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f1617c2353aed6f23004b383ad3a8800e33a8e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 16:30:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1651e000bc444e848fdb66e0b6ef1f00 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11961f6ec34f4d33989cc451e9dc3ecd : SUCCESS in 4m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bdf9bc14cc0418d83186226f3be2148 : SUCCESS in 7m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d52a8538413643f49cb261b71ecde820 : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dd93bf73a324769a56885e433487d2a : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2174b7b5f9ac48d0b034ea5ceaa60784 : SUCCESS in 6m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6ebc050dffe40d59964c3de6b200608 : SUCCESS in 4m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b723adf7e0204defb46b750de5a1bf40 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b0bd9f96ba7472ab0171200c2b0b263 : SUCCESS in 7m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4bff65700c4e483e91e55ca031684216 : SUCCESS in 28m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/86a8a0d8c44f4a38817e83d1a612c9c7 : SUCCESS in 57m 10s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/fb120b86faa9410f83dd301ca89d51c8 : SUCCESS in 1h 32m 32s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b6c2ebccf0f6401b8b4dd6aabd4bbdf1 : SUCCESS in 20m 31s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b53d378ad8d94aa3ada814daf05ea48a : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":1},{"id":"352943a14501edc369c26e2f1a1c1830ac68dd75","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-01 16:40:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"49c829eb02564a883dfaeedb1368e3cb957f863a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 16:40:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57425327b746f764a604a254f6377e8b96f69f5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 18:15:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/1b2b77e3e0a841af8c01c3924a32ddba : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/922ddebf39024307a7860d099229a65f : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e591ad24865343ef9cfe3bef5b5b4e79 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c6c35ecc314850a50ce68bb44e4cda : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7d74d8b14e134f67879cc5700fff6e93 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec868847dc044e6cb19f3f6c3f3115cc : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92c74d5fa2584726961b2afb1568ee7e : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/106d49f0aabe4022ad24bb0b2660d1cc : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/44d3f0d82f1645229bbbf23b2d1897de : SUCCESS in 33m 23s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/cc966a145bb545879fa140c8ab94f607 : SUCCESS in 1h 08m 59s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/49a6b60137be4d2ba5204c28d99489d4 : SUCCESS in 1h 31m 07s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/69d61b72b4dd4d0ebe6a733684d73619 : SUCCESS in 19m 04s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/24fd0c9513a741deb241dcc3adcbd3e6 : SUCCESS in 39m 29s","accounts_in_message":[],"_revision_number":1},{"id":"390231410a031637afa585c9bd58d00c1e84cc6c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-01 18:15:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d90bdaeb4466da58dec39db1e2dfeaf01cffb174","revisions":{"d90bdaeb4466da58dec39db1e2dfeaf01cffb174":{"kind":"REWORK","_number":1,"created":"2019-10-31 11:14:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/36/692336/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/692336/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/692336/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/692336/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/692336/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/692336/1"}}},"commit":{"parents":[{"commit":"852f7566699d9074d2f830589f9eeab78fd0fc0e","subject":"Merge \"Set provisioning and cleaning networks as non-shared\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/852f7566699d9074d2f830589f9eeab78fd0fc0e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-04 15:49:17.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-31 11:14:07.000000000","tz":0},"subject":"Fix baremetal serial console autoenable","message":"Fix baremetal serial console autoenable\n\nIf ironic_serial_console_autoenable is set to true, ironic compute node\nserial consoles are automatically enabled during kayobe overcloud post\nconfigure. However, the baremetal-compute-serial-console.yml playbook\nexpects a variable called cmd to be defined, and set to \u0027enable\u0027 in\norder to enable serial consoles. This patch fixes the issue.\n\nWe also add a conditional to only import the playbook when ironic is\nenabled in the overcloud.\n\nChange-Id: Ice26a113f9023b1a25a4af0552d07c5e95e64556\nStory: 2006662\nTask: 36924\n(cherry picked from commit 5977a4411921c1880ee023b945ac2cceb533749a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d90bdaeb4466da58dec39db1e2dfeaf01cffb174"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d90bdaeb4466da58dec39db1e2dfeaf01cffb174"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
