)]}'
{"id":"openstack%2Fkayobe~692824","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Iaed6320f23a4dfe7a69e2921d363bf3bdcf035ef","project":"openstack/kayobe","branch":"stable/stein","topic":"kolla-venv-stable/stein","hashtags":[],"change_id":"Iaed6320f23a4dfe7a69e2921d363bf3bdcf035ef","subject":"Only set up the kolla ansible venv on hosts in the kolla ansible inventory","status":"MERGED","created":"2019-11-04 14:25:42.000000000","updated":"2019-11-14 14:39:20.000000000","submitted":"2019-11-14 14:37:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692824-1573742226191-1353a74c","meta_rev_id":"5aa65eac0f1982fb0ab397b307ad3d028d081c20","_number":692824,"virtual_id_number":692824,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 14:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-11-14 09:53:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-14 12:31:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-14 12:31:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-11-04 14:25:42.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 14:25:42.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-14 12:31:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-14 14:37:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"291af51c864cfca5c28d19809a4c3389cc6c146e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:25:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"545b3d334e77f18e41f04ba6ee7c77bef322af66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:23:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94593615557c4aff918af378081c88a7 : SUCCESS in 5m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ce321fd6d7d47e6bddaa8fca3b5835c : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66bd0d7d8d024c3a9e749ac816119541 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c30c8394411f4b3792fad3f470093ba2 : SUCCESS in 5m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/73753e1af2544667b2900470605add4e : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e46f0f2a3689461ca8918637745cca85 : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2884c9da5f8745e68773d0dd2ae0d7c0 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0c07477529e549fca0895d9eef4c2018 : SUCCESS in 9m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5b865b4f0174939b591a8d4445be5cc : SUCCESS in 40m 43s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4ab9c8c4f8fa4e2b85a6980f12dd0bcc : SUCCESS in 58m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/251e08a7d405448b90445a168c5f045c : SUCCESS in 1h 56m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1131ccd3959649bda6347d0ee44866f8 : SUCCESS in 22m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4962118b776243829f8fa0e6c138b6c1 : SUCCESS in 39m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5f3d124f8fcea67283899ee6f2b863f81880831a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-05 11:17:55.000000000","message":"Patch Set 1: Code-Review-1\n\nWhile looking at this change again, I noticed that the default host limit is \"seed:overcloud\". With this patch the playbook will never run on the seed?\n\nAlso, kolla-target-venv is called by SeedHypervisorHostUpgrade, but the playbook is limited to seed:overcloud (that\u0027s probably an issue with the code in SeedHypervisorHostUpgrade).","accounts_in_message":[],"_revision_number":1},{"id":"2724490171e79b814318d79b35606a2693530ab5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 11:39:59.000000000","message":"Patch Set 1:\n\n\u003e While looking at this change again, I noticed that the default host\n \u003e limit is \"seed:overcloud\". With this patch the playbook will never\n \u003e run on the seed?\n \u003e \n \u003e Also, kolla-target-venv is called by SeedHypervisorHostUpgrade, but\n \u003e the playbook is limited to seed:overcloud (that\u0027s probably an issue\n \u003e with the code in SeedHypervisorHostUpgrade).\n\nGood catch @pierre. Looks like we didn\u0027t catch it due to the AIO test environment? So we need to add the seed group to this conditional?","accounts_in_message":[],"_revision_number":1},{"id":"939a8c34d87d21816eaeb46f9bc46e21d19eb2bd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-13 14:13:18.000000000","message":"Patch Set 1:\n\nShould we try to merge this followed closely by a backport of https://review.opendev.org/#/c/693030/","accounts_in_message":[],"_revision_number":1},{"id":"ecb757ba7cd2f576c88392a72c73585b487fc3c7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 09:53:26.000000000","message":"Patch Set 1:\n\n\u003e Should we try to merge this followed closely by a backport of\n \u003e https://review.opendev.org/#/c/693030/\n\nProposed: https://review.opendev.org/#/c/694220","accounts_in_message":[],"_revision_number":1},{"id":"7025fa8f4bf4d43c727c6cbb3e85adeb9023a882","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 09:53:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d818f9ce5d7faa49c5ee25fa1ff3d1ce24c6f86","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-14 12:31:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d705f5b3d48e6efec523cf30bf2974e9ead96cef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 12:32:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7769a0f133ee9646f5cdb094b26d950a04968204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:37:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/af3c75719ac246dbb5d629ca33406445 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f14298759af4d7b8ace8ed910344d29 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b3ef07c7ec4922b867442821ec3982 : SUCCESS in 5m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f457cec1485a405281525136b90aca98 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c140c2225a431ebef94752c81ab1fb : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f40b00ad27ee4ebf9ea5a41fc66c5e8b : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c41016b4bbbd4b7aad9f8bf98d07b66f : SUCCESS in 8m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f07e47f4c9a445ea7593e18d75896ad : SUCCESS in 38m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3bae2e39048a4be8af1215d90fd55b8c : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3c8f65f0d9cd469c801dc60420148c84 : SUCCESS in 1h 46m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ae555d8af1cf4d318c55ab1c36d6556f : SUCCESS in 20m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/18ddf3e2c03945348322ad0cea9a8e28 : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":1},{"id":"4f643e87a9534c869f1384b3b568b1c9dba3523b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:37:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"97b62ad38dddcc13e43ef8c8dab7b0294663b842","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:39:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/944a9841a96546a6a200559438a9f921 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f40b64ebd82b5179a1794a80874060f23fdcb813","revisions":{"f40b64ebd82b5179a1794a80874060f23fdcb813":{"kind":"REWORK","_number":1,"created":"2019-11-04 14:25:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/24/692824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/692824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/692824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/692824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/692824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/692824/1"}}},"commit":{"parents":[{"commit":"a5ba5dc0e31d09cbb16361d5e49e7b7adc5b5380","subject":"Merge \"Example should use hyphen instead of underscore\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5ba5dc0e31d09cbb16361d5e49e7b7adc5b5380"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-07-18 14:17:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 14:25:42.000000000","tz":0},"subject":"Only set up the kolla ansible venv on hosts in the kolla ansible inventory","message":"Only set up the kolla ansible venv on hosts in the kolla ansible inventory\n\nChange-Id: Iaed6320f23a4dfe7a69e2921d363bf3bdcf035ef\nStory: 2006247\nTask: 35856\n(cherry picked from commit 300f90134e7d4bdc298cc050d60f67476145c87b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f40b64ebd82b5179a1794a80874060f23fdcb813"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f40b64ebd82b5179a1794a80874060f23fdcb813"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
