)]}'
{"id":"openstack%2Fkayobe~692825","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Iaed6320f23a4dfe7a69e2921d363bf3bdcf035ef","project":"openstack/kayobe","branch":"stable/rocky","topic":"kolla-venv-stable/rocky","hashtags":[],"change_id":"Iaed6320f23a4dfe7a69e2921d363bf3bdcf035ef","subject":"Only set up the kolla ansible venv on hosts in the kolla ansible inventory","status":"MERGED","created":"2019-11-04 14:25:53.000000000","updated":"2019-11-14 15:04:29.000000000","submitted":"2019-11-14 15:04:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692825-1573743869105-bdb6f180","meta_rev_id":"f9493e4bffe0f75656f008fa173712551ab3c207","_number":692825,"virtual_id_number":692825,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 15:04:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-11-14 09:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-14 13:17:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-14 13:17:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-11-04 14:25:53.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 14:25:53.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-14 13:17:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-14 15:04:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b4459323b2cff9d8e46697be389827ace8c3ddd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:25:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dddfbcc4df7065faff4caafb7babf731007f37ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 15:59:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b003226c3cc34656bdaaa9df55b12942 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d0bf4453904696bf7e423179c21dd8 : SUCCESS in 6m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bda071917d7946bb86f6732a4474a0ac : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1a760de503cc4efcb0b7fed69f2173f0 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34addcbb06894fea8c10eefee2a14607 : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec050ad598814831a012006c3a061685 : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/099c4ea319c04632a3bc5a82fb920531 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9e4725bcb07344279a4f988f1aedfc22 : SUCCESS in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e37febe1c9b4d5ca5cc138388b4c0be : SUCCESS in 36m 49s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/358c5ba0cdeb4d1ca7de908b468de57e : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b6238dca95414e8697f803e9bfdc78a2 : SUCCESS in 1h 31m 56s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/a505ca526ae840e5b5abb874cb201116 : SUCCESS in 17m 16s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/da6d6088cf5b4fa1b5a85f389c12684f : SUCCESS in 42m 03s","accounts_in_message":[],"_revision_number":1},{"id":"14f37a01bbf3c98172f8b43057af444577b8c52a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 09:57:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b97121a397b5a65e8c7013e6ed01397f1d1a9eca","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-14 13:17:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8c2e9b0fc51138c3cb2698b3b7636e4d7cf0f950","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 13:17:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2d42ec28b471c188f442d7b846a70e3a0f087c8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 15:04:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2de8a186141543b8a7ba01a1d76525b0 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9442b3101aa941ecb2879a649f7a489a : SUCCESS in 3m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a707296d057246299263429c46a4a6d8 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3affe08df427435c9319398d265bd3b6 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8319dcc778134903a948fa9ef7ce8521 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/221c9c2c83004a8ea6376dd28124bd80 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b09be96fb7eb47e5abeb19ca5107fdc3 : SUCCESS in 8m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e3b79ddaad4747c6b87db29491402131 : SUCCESS in 27m 03s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/00059ba7e24149d5a0a537e05941d2cf : SUCCESS in 59m 46s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2934f48b4a924c35ad1d40e14d7a4baa : SUCCESS in 1h 45m 56s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/14dee08d7b484e5a8987eaa7bb3bdfa6 : SUCCESS in 16m 50s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/fb9d0bb85b6241508cac56dfe22bb6b0 : SUCCESS in 32m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9f95fb021b1d6c49b4e0beda13ae43f2d7e43a40","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 15:04:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"569e7871d4b01576ad6439b6aaa39a00187ba0ad","revisions":{"569e7871d4b01576ad6439b6aaa39a00187ba0ad":{"kind":"REWORK","_number":1,"created":"2019-11-04 14:25:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/25/692825/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/25/692825/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/692825/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/692825/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/25/692825/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/25/692825/1"}}},"commit":{"parents":[{"commit":"0865b4242b5bee55e398436ab4ae7793321e3fca","subject":"Merge \"CI: Don\u0027t set cpu mode\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0865b4242b5bee55e398436ab4ae7793321e3fca"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-07-18 14:17:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 14:25:53.000000000","tz":0},"subject":"Only set up the kolla ansible venv on hosts in the kolla ansible inventory","message":"Only set up the kolla ansible venv on hosts in the kolla ansible inventory\n\nChange-Id: Iaed6320f23a4dfe7a69e2921d363bf3bdcf035ef\nStory: 2006247\nTask: 35856\n(cherry picked from commit 300f90134e7d4bdc298cc050d60f67476145c87b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/569e7871d4b01576ad6439b6aaa39a00187ba0ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/569e7871d4b01576ad6439b6aaa39a00187ba0ad"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
