)]}'
{"id":"openstack%2Fkayobe~692827","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I61067164a003bcdf6079cdaca36f0e1755186f60","project":"openstack/kayobe","branch":"stable/stein","topic":"story/2006669-stable/stein","hashtags":[],"change_id":"I61067164a003bcdf6079cdaca36f0e1755186f60","subject":"Fix validation of locally built IPA when ironic is disabled","status":"MERGED","created":"2019-11-04 14:27:40.000000000","updated":"2019-11-25 17:47:49.000000000","submitted":"2019-11-25 17:43:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692827-1574703819878-8eeca06d","meta_rev_id":"3098e5f15ee389f1f639b889a6514d086d86010c","_number":692827,"virtual_id_number":692827,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 17:43:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-04 15:42:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-11-25 15:44:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 15:42:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-25 15:44:55.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-11-25 17:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4425c939b3a9d1a9969d75fd123576e36c41c13f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:27:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1820542fd7db8816a5cda44959636b64e5082f83","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-04 15:42:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f69061925a3d0cca103134d5eed7ca0ab3e60707","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 15:58:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e43caa85517a4c72967db4011302d268 : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/835ea1689a014855aa269aa4d72a43b3 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfa396b5e4c043b78b3c853ebeee4030 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e55d6a47a924776af42af88b68df1cd : SUCCESS in 4m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d2b5f60d8bb48bfbbea357210241f93 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e83163fbcf4e7eac336ca8a9d05710 : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b9ae72f62b34f1c91a6433276cefa42 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f692998dec074dbaa3ddd0c218b18e88 : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/42f388e99bfd4df9ba3ffa77120f664a : SUCCESS in 30m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8195749c1b7e4570b90a9d30a690db4b : SUCCESS in 1h 08m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ae0ba26db0784b76a2b372c359870564 : SUCCESS in 1h 29m 30s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/276b58a27eb14bffa8392b0e5a80363a : SUCCESS in 26m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/6a4101e5fa0d449ba2a1e75f02fffae9 : SUCCESS in 34m 23s","accounts_in_message":[],"_revision_number":1},{"id":"15d939832497c4148fbad396dd0a4bf33c5dd789","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-25 15:44:55.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6b38fd775f6420ae8a80f4830aa041f1ff6d155e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:45:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93778cd21b33df837a4a20bf6a65708ab7e076e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:43:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"57a64a5b0b94d35f578dd95112aadefa61170e9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:43:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fde869f5ed8c40d2b9ef1f0224b5259c : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c70bc81a9b84e17894df8c20c3860ea : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90300a312b4d4b12b7841c3f32e4a985 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf67dfa343864599aa46f6f8203f009a : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/410938f11a394f00bac57e7561982be0 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b2d8927d2e143eeb901dccce535689e : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fb43accddbe545f58b46943650ace0fd : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b842f8906203478589f77196f7005a81 : SUCCESS in 36m 58s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/695b5013ee6e46e49e45d492e8fe63dd : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3141b8e22a264d42962c35da1559d919 : SUCCESS in 1h 35m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ff749811f17d4135a834fc3e48f7b86d : SUCCESS in 20m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/42d19aed61b5497691d2da5a67f5a6a4 : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":1},{"id":"abe3e6fc47be1dab0b651b3c40eaa4143861a184","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:47:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7ade015e1574241a6730a6a85b516b9 : SUCCESS in 1m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08f093e369c002e8f071006d4e03b7f673afaeac","revisions":{"08f093e369c002e8f071006d4e03b7f673afaeac":{"kind":"REWORK","_number":1,"created":"2019-11-04 14:27:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/27/692827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/692827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/692827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/692827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/692827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/692827/1"}}},"commit":{"parents":[{"commit":"a5ba5dc0e31d09cbb16361d5e49e7b7adc5b5380","subject":"Merge \"Example should use hyphen instead of underscore\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5ba5dc0e31d09cbb16361d5e49e7b7adc5b5380"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-07 12:50:35.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 14:27:40.000000000","tz":0},"subject":"Fix validation of locally built IPA when ironic is disabled","message":"Fix validation of locally built IPA when ironic is disabled\n\nIn an environment where ironic is disabled, but IPA images are built\nlocally for the seed, the following task will fail in a number of\ncommands, e.g. kayobe overcloud service deploy:\n\nValidate the presence of locally built Ironic Python Agent (IPA) images\n\nWe need to only execute this task when ironic is enabled.\n\nChange-Id: I61067164a003bcdf6079cdaca36f0e1755186f60\nStory: 2006669\nTask: 36935\n(cherry picked from commit b05fc00dcf059ce961eb1355b359b443883cb17f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08f093e369c002e8f071006d4e03b7f673afaeac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08f093e369c002e8f071006d4e03b7f673afaeac"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
