)]}'
{"id":"openstack%2Fkayobe~692828","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I61067164a003bcdf6079cdaca36f0e1755186f60","project":"openstack/kayobe","branch":"stable/rocky","topic":"story/2006669-stable/rocky","hashtags":[],"change_id":"I61067164a003bcdf6079cdaca36f0e1755186f60","subject":"Fix validation of locally built IPA when ironic is disabled","status":"MERGED","created":"2019-11-04 14:27:50.000000000","updated":"2019-11-25 17:43:25.000000000","submitted":"2019-11-25 17:43:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692828-1574703805695-8eeca06d","meta_rev_id":"54a051d8faece41f1861aa335239683348e214c6","_number":692828,"virtual_id_number":692828,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 17:43:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-04 15:42:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-11-25 15:39:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 15:42:41.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-25 15:39:32.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-11-25 17:43:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0e19d8baf85340c15b6f54ec1713289fffb7a3e8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:27:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b3c698316e86ea09fba16a37d0696373a5c7546f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-04 15:42:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f2dbb2d4ce4f75e32a2e7d33ac46a2664db564e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:17:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4f71e91b590b4005a4442bb242a16d7b : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/767acde2bf87493a9f6d24b6975cf181 : SUCCESS in 3m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbd3efeb1c4c412c95d83abcc5683862 : SUCCESS in 3m 40s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8944b5557284449998c7a6adea56a2e8 : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed14a9c87db4f65a5617dd1ef251880 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6e9f27d8b282411b81e9b5b1a5827925 : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c77992cb61884034a7b5dc58624191e9 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/98bcad2767a74e84a174c423f603e6df : SUCCESS in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/23a364273fbb420e80cb09dc2a853d60 : SUCCESS in 31m 06s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3bb8642750164fca815b63ad47e3e50d : SUCCESS in 1h 02m 33s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f64fb1a5fce542c6b43b47b905066f1d : SUCCESS in 1h 47m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3855e7c05a2c436895ddd35c9d01466b : SUCCESS in 24m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/98f0e6b6bb4844a19498ef070893b7a5 : SUCCESS in 39m 49s","accounts_in_message":[],"_revision_number":1},{"id":"890569530a90d435757228d036a9b475c63e8f0f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-25 15:39:32.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e5a4bb6af274dc7736470ea2e94383ea60e2325a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:39:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"773fcd8704e78802fe3b3e5310df7ac4f5a0301a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:43:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7bb5806106de4eb2a958e74299a6dbbd : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a2a8bc13254267bd8a2f7d64d4e560 : SUCCESS in 3m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/094a884ad2d0427abfa99654f024ba28 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e263ff6a96f40da9df8bf3e2142fc51 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/da73e16c807c42a2a96e9c97ae426c4d : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d2594cc752574e998a13e5b536613610 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3313d9588ede4b978c89c020af6885e7 : SUCCESS in 7m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3fd664fb39544f5eb485e94c57e7be76 : SUCCESS in 33m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d3a1b5ac09ab428c904f552f0547b04e : SUCCESS in 55m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/580729f0415944d6853d187adb01026e : SUCCESS in 2h 02m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/bb20f9ec26f44a0ea901d88467f37812 : SUCCESS in 18m 47s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a6e38f21bb0b466fb84b334674ebda5c : SUCCESS in 33m 55s","accounts_in_message":[],"_revision_number":1},{"id":"4744ea2a5f09523b71f651a5212af90f72b6bfce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:43:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec6a6e39c36a3f71e337dd48c0888b883d9b715c","revisions":{"ec6a6e39c36a3f71e337dd48c0888b883d9b715c":{"kind":"REWORK","_number":1,"created":"2019-11-04 14:27:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/692828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/28/692828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/692828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/692828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/28/692828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/28/692828/1"}}},"commit":{"parents":[{"commit":"0865b4242b5bee55e398436ab4ae7793321e3fca","subject":"Merge \"CI: Don\u0027t set cpu mode\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0865b4242b5bee55e398436ab4ae7793321e3fca"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-07 12:50:35.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 14:27:50.000000000","tz":0},"subject":"Fix validation of locally built IPA when ironic is disabled","message":"Fix validation of locally built IPA when ironic is disabled\n\nIn an environment where ironic is disabled, but IPA images are built\nlocally for the seed, the following task will fail in a number of\ncommands, e.g. kayobe overcloud service deploy:\n\nValidate the presence of locally built Ironic Python Agent (IPA) images\n\nWe need to only execute this task when ironic is enabled.\n\nChange-Id: I61067164a003bcdf6079cdaca36f0e1755186f60\nStory: 2006669\nTask: 36935\n(cherry picked from commit b05fc00dcf059ce961eb1355b359b443883cb17f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec6a6e39c36a3f71e337dd48c0888b883d9b715c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec6a6e39c36a3f71e337dd48c0888b883d9b715c"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
