)]}'
{"id":"openstack%2Fkayobe~692830","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05","project":"openstack/kayobe","branch":"stable/stein","topic":"dual2-stable/stein","hashtags":[],"change_id":"I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05","subject":"Fix issue with image download when checksum url is none","status":"MERGED","created":"2019-11-04 14:28:44.000000000","updated":"2019-11-25 17:45:42.000000000","submitted":"2019-11-25 17:43:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692830-1574703809220-1353a74c","meta_rev_id":"6a1d6fd2a6e577785d154b171cb6874ffa5bd7d0","_number":692830,"virtual_id_number":692830,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 17:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-04 15:43:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-11-25 15:39:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-11-04 14:28:44.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 14:28:44.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-04 15:43:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-25 15:39:48.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-11-25 17:43:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ae065b56f268774c7ad669d2799b9ecc183fb6c8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:28:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"330a2d8e6c043c2fba57baa3e2f2c0d5dd67c841","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-04 15:43:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04a5cbd06799d9f37f8a1f606d4000addb963bac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:16:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b22b88b0cf794972adffc1e87d3d2d96 : SUCCESS in 4m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9f42416cca30414d82d207f53c75c445 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63cecdb4358a49c393c882a72da3e57c : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d396d503a0a44f309579f6436e687788 : SUCCESS in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0fb278bcd28743e6902478ce3ff5132a : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86ff1450150246ac9f2bef9c0d13d884 : SUCCESS in 3m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9921843aa8a4d9a8ace7ea92de0a6c0 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75fffc58287c4c1eb65d3ed4ce23eca4 : SUCCESS in 7m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a5a563ecddd5437a9e985611dc70c49c : SUCCESS in 29m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9ac908e2a7e0494bb059352dffd69751 : SUCCESS in 54m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/2d2c6300bc3a4f1fa8605e02dcbe7c43 : SUCCESS in 1h 42m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/998ddbfcd55c4d60b4c69700966f9f9b : SUCCESS in 18m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/edd1d1f7325f4fe0b228446c0a13f3d2 : SUCCESS in 29m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a22b7acf606ce045d7fb0934ec02bfac6c3ffa4f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-25 15:39:48.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8004bfb76ac1942572f09cea9dbf53c981274796","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:40:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7d1f178a10efcb82d0579d16a8d7ba66ab07b254","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:43:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2a3e6fee8d0d5339ec454cc68c4f6cc2d312744c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:43:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/103ab086a41647ada0e818674777421f : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0162e4dec84641ba8e44defcb89430d2 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c5d65f4151d462d85ecf50f0a25292d : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd4f6457c59d4970aa6e74442c15729c : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/095ea7e4d0dc4e898d38e1fc12e503f9 : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68a229ac27994aedada7e8ec6ed3f2e5 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/736d997a7eea4adcb6ba1c2fb104d4e8 : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/509d2683062c4b2981c3db449902bac7 : SUCCESS in 41m 18s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5bdcf33d30e74d1bafac11a8b36f4299 : SUCCESS in 1h 01m 49s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/312227fdbe8c4b55840cf3b0756402a3 : SUCCESS in 1h 56m 47s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/10910b93b34848979466abfc59d34261 : SUCCESS in 26m 32s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/abbd118ef35d4c37a2f6a352f4885be8 : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":1},{"id":"09b52c943979756dffb6d6088ecb2114ad90dab7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:45:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02481c27c1cb440191ce760073f32228 : SUCCESS in 1m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e1b92436137181d560a5f21570fb9e44114cfad","revisions":{"8e1b92436137181d560a5f21570fb9e44114cfad":{"kind":"REWORK","_number":1,"created":"2019-11-04 14:28:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/692830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/692830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/692830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/692830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/692830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/692830/1"}}},"commit":{"parents":[{"commit":"a5ba5dc0e31d09cbb16361d5e49e7b7adc5b5380","subject":"Merge \"Example should use hyphen instead of underscore\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5ba5dc0e31d09cbb16361d5e49e7b7adc5b5380"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 09:17:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 14:28:44.000000000","tz":0},"subject":"Fix issue with image download when checksum url is none","message":"Fix issue with image download when checksum url is none\n\nTASK [image-download : Ensure the image is downloaded] *************************\ntask path: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml:19\nfatal: [localhost]: FAILED! \u003d\u003e {\n    \"msg\": \"The field \u0027vars\u0027 has an invalid value, which includes an undefined variable. The error was: \u0027dict object\u0027 has no attribute \u0027content\u0027\\n\\nThe error appears to have been in \u0027/home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml\u0027: line 19, column 7, but may\\nbe elsewhere in the file depending on the exact syntax problem.\\n\\nThe offending line appears to be:\\n\\n\\n    - name: Ensure the image is downloaded\\n      ^ here\\n\"\n}\n\tto retry, use: --limit @/home/zuul/kayobe-venv/share/kayobe/ansible/kolla-openstack.retry\n\nPLAY RECAP *********************************************************************\ncontroller0                : ok\u003d3    changed\u003d0    unreachable\u003d0    failed\u003d0\nlocalhost                  : ok\u003d6    changed\u003d2    unreachable\u003d0    failed\u003d1\n\nChange-Id: I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05\nStory: 2006715\nTask: 37086\n(cherry picked from commit dea36ac9d24c1e8cf19f0d27ee49be657d736da6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e1b92436137181d560a5f21570fb9e44114cfad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e1b92436137181d560a5f21570fb9e44114cfad"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
