)]}'
{"id":"openstack%2Fkayobe~692831","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05","project":"openstack/kayobe","branch":"stable/rocky","topic":"dual2-stable/rocky","hashtags":[],"change_id":"I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05","subject":"Fix issue with image download when checksum url is none","status":"MERGED","created":"2019-11-04 14:28:59.000000000","updated":"2019-11-25 17:29:53.000000000","submitted":"2019-11-25 17:29:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692831-1574702993250-7789a045","meta_rev_id":"9c148d93d7ab1fbe59724c3de9f18571032d3e49","_number":692831,"virtual_id_number":692831,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 17:29:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-04 15:43:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-11-25 15:38:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-11-04 14:28:59.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 14:28:59.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-04 15:43:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-25 15:38:58.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-11-25 17:29:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b449f4583bd852cb4406bea261b3efb599cc6a43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:28:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"57906f813a8c38c810c1c8ec6b3f630a2bc9b6c9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-04 15:43:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"316f69b1f2ac88418c854d6e68a23beea8c15ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 16:24:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/7036d3a58a374b05a0a44ea4ddb88661 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097945f026ff48d4bcabdd162da57380 : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3309aef41e8a4bd88a6a754e26136710 : SUCCESS in 5m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/90aae8bc5a534eefa69b57c74f37a7a0 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b4903a1cc245208f75c790922447d3 : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa751e2584a743068e5d7f219eee0c9e : SUCCESS in 6m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f04ae64c3e5441da890516bc0e142833 : SUCCESS in 5m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/07ae9717fb9048899b4ac4a0614532d4 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ab940f03b0a34ca2a13057c8664ca6d1 : SUCCESS in 35m 02s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5bf8e1aba5444d7da791881aebfcc486 : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/6c645866eed14c3a82aebe0623573d65 : SUCCESS in 1h 54m 43s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4884a6b513de431a9d9d4d6c628ef06a : SUCCESS in 18m 50s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7aac428a3243483d9276b7b8c9607c44 : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b7c2fa2092d18b4f88dd5d76d9d05cea47014197","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-25 15:38:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"032be705de61c64949d94a8ba1237303de62fad9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:39:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1655e4ab6de8e87dad29250d16aee84a8816aa0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:29:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c83f7e9f6e10bbfce045280771c5e7c73887142f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:29:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f027231e57444bf493e7139bd991da85 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef42dc41d448408591b603166055b8e2 : SUCCESS in 3m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d576e53a3d20430dae05bd9adf8c0e90 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01861442d74645128d54bd8684691522 : SUCCESS in 3m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3c978290e9864ef9ad57e6717ec5d161 : SUCCESS in 5m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/38da500e022f416ebd085753bbd65d27 : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75c8a683e5be4d5ca5afcb8faa113076 : SUCCESS in 8m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/56d8fd69dace4b75ba35ba53c7b95cfb : SUCCESS in 25m 19s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9c3607e8aed54a579b56140bb82ffd79 : SUCCESS in 1h 13m 14s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f64033a8d4b64a6f9d06af253e278f0a : SUCCESS in 1h 48m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f72b412298404e83b0b61708bfa84783 : SUCCESS in 20m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4c911b4a92c74e9fb6e4331a5c22fc2d : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4ea80760c2dd8f1bcc4c8079e379f721f00d242","revisions":{"d4ea80760c2dd8f1bcc4c8079e379f721f00d242":{"kind":"REWORK","_number":1,"created":"2019-11-04 14:28:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/692831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/692831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/692831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/692831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/692831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/692831/1"}}},"commit":{"parents":[{"commit":"0865b4242b5bee55e398436ab4ae7793321e3fca","subject":"Merge \"CI: Don\u0027t set cpu mode\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0865b4242b5bee55e398436ab4ae7793321e3fca"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 09:17:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 14:28:59.000000000","tz":0},"subject":"Fix issue with image download when checksum url is none","message":"Fix issue with image download when checksum url is none\n\nTASK [image-download : Ensure the image is downloaded] *************************\ntask path: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml:19\nfatal: [localhost]: FAILED! \u003d\u003e {\n    \"msg\": \"The field \u0027vars\u0027 has an invalid value, which includes an undefined variable. The error was: \u0027dict object\u0027 has no attribute \u0027content\u0027\\n\\nThe error appears to have been in \u0027/home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml\u0027: line 19, column 7, but may\\nbe elsewhere in the file depending on the exact syntax problem.\\n\\nThe offending line appears to be:\\n\\n\\n    - name: Ensure the image is downloaded\\n      ^ here\\n\"\n}\n\tto retry, use: --limit @/home/zuul/kayobe-venv/share/kayobe/ansible/kolla-openstack.retry\n\nPLAY RECAP *********************************************************************\ncontroller0                : ok\u003d3    changed\u003d0    unreachable\u003d0    failed\u003d0\nlocalhost                  : ok\u003d6    changed\u003d2    unreachable\u003d0    failed\u003d1\n\nChange-Id: I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05\nStory: 2006715\nTask: 37086\n(cherry picked from commit dea36ac9d24c1e8cf19f0d27ee49be657d736da6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4ea80760c2dd8f1bcc4c8079e379f721f00d242"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d4ea80760c2dd8f1bcc4c8079e379f721f00d242"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
