)]}'
{"id":"openstack%2Fkayobe~692832","triplet_id":"openstack%2Fkayobe~stable%2Fqueens~I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05","project":"openstack/kayobe","branch":"stable/queens","topic":"dual2-stable/queens","hashtags":[],"change_id":"I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05","subject":"Fix issue with image download when checksum url is none","status":"MERGED","created":"2019-11-04 14:29:11.000000000","updated":"2019-11-25 17:43:32.000000000","submitted":"2019-11-25 17:43:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692832-1574703812908-c4d0231e","meta_rev_id":"d4a22b2d01bf4c98c124ad5d220ad1eff868be34","_number":692832,"virtual_id_number":692832,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 17:43:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-04 15:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-11-25 15:40:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-11-04 14:29:11.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-04 14:29:11.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-04 15:43:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-25 15:40:14.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-11-25 17:43:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"099dcb36235d4c185075b07ca9ec160e0b689fdd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-04 14:29:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"86ce32963a6af383210da30b7507fbd1c2d0874a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-04 15:43:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2de09ae03e7342ec19565e0d9d04d174f8818ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-04 15:54:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/19c0a53e28104a95ac34826cfb7c715e : SUCCESS in 5m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c4dd78f9fac4a558f4f36aff6eeb386 : SUCCESS in 3m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9555cb01e2b34e9a99480bab097d0d60 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d020db3246df44959558536ab157aba4 : SUCCESS in 6m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ac53163fe1ff4f0c8fc9b4be44f21433 : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cda5579696d44e7eaf177af8fd2edd1a : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f173c7c419e49a991578cde70904c82 : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0823ee3d598843ac9d5d762a93ca1abe : SUCCESS in 31m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d78b0712e76342ce8b13dd3d8a29cfca : SUCCESS in 48m 17s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/819936c005b7469b89b935439de89f1e : SUCCESS in 1h 25m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2e0612f437314f8ba0323bd5dd778202 : SUCCESS in 24m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d5d844e9051044068c4bc7c308762c61 : SUCCESS in 37m 40s","accounts_in_message":[],"_revision_number":1},{"id":"8dd4c5ee1fa63e05d61491ea32f2a2934cf5ee41","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-25 15:40:14.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a13b5a20c4988406f4b69a7517a0a81b251ac995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:40:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"827aeb934f3813494eadcd1a49bfdf842d5f832c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:43:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c08afecccb7f7c8c54aebf78c6108f544e3bd502","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 17:43:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/70656e89b3074f7eba135f9da605d8af : SUCCESS in 3m 57s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e6d4fb7d941d410181fe7434154a3c94 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df5004d6a05e489f8cecfc0cab9b8905 : SUCCESS in 4m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c80dce73f1154761a2f2a8851eb7a3ec : SUCCESS in 3m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c94708a08acf4f9d8fba671df504faa6 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0787637ac8454cb9a4b3f8187c33a1e9 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/994c902b981d43fda0f99c15cdd1d839 : SUCCESS in 25m 03s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/05af280701084bbcbe0827552235478f : SUCCESS in 51m 46s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/dd796cf44fcd4d4c8b640d67d15667c7 : SUCCESS in 1h 36m 01s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/62aef5db6d69407eaaaf31f8851154b8 : SUCCESS in 20m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4cbc5b76021b4e34a507e01379afb907 : SUCCESS in 44m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b62ec97cd43428470f88491a835054321d8730d4","revisions":{"b62ec97cd43428470f88491a835054321d8730d4":{"kind":"REWORK","_number":1,"created":"2019-11-04 14:29:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/692832/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/692832/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/692832/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/692832/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/692832/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/692832/1"}}},"commit":{"parents":[{"commit":"fe7cfed080037b0ad473442645b8cad3b23c3228","subject":"Use HTTPS for cirros download","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fe7cfed080037b0ad473442645b8cad3b23c3228"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-14 09:17:47.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-04 14:29:11.000000000","tz":0},"subject":"Fix issue with image download when checksum url is none","message":"Fix issue with image download when checksum url is none\n\nTASK [image-download : Ensure the image is downloaded] *************************\ntask path: /home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml:19\nfatal: [localhost]: FAILED! \u003d\u003e {\n    \"msg\": \"The field \u0027vars\u0027 has an invalid value, which includes an undefined variable. The error was: \u0027dict object\u0027 has no attribute \u0027content\u0027\\n\\nThe error appears to have been in \u0027/home/zuul/kayobe-venv/share/kayobe/ansible/roles/image-download/tasks/main.yml\u0027: line 19, column 7, but may\\nbe elsewhere in the file depending on the exact syntax problem.\\n\\nThe offending line appears to be:\\n\\n\\n    - name: Ensure the image is downloaded\\n      ^ here\\n\"\n}\n\tto retry, use: --limit @/home/zuul/kayobe-venv/share/kayobe/ansible/kolla-openstack.retry\n\nPLAY RECAP *********************************************************************\ncontroller0                : ok\u003d3    changed\u003d0    unreachable\u003d0    failed\u003d0\nlocalhost                  : ok\u003d6    changed\u003d2    unreachable\u003d0    failed\u003d1\n\nChange-Id: I45fb193b423c7e874aa70cd5f0252e0b0f4e2f05\nStory: 2006715\nTask: 37086\n(cherry picked from commit dea36ac9d24c1e8cf19f0d27ee49be657d736da6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b62ec97cd43428470f88491a835054321d8730d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b62ec97cd43428470f88491a835054321d8730d4"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
