)]}'
{"id":"openstack%2Fkayobe~692981","triplet_id":"openstack%2Fkayobe~master~I02adfa5819bff781e1649b81c1023b55de32ebd0","project":"openstack/kayobe","branch":"master","topic":"host-command-facts","hashtags":[],"change_id":"I02adfa5819bff781e1649b81c1023b55de32ebd0","subject":"Stop gathering facts for host command run","status":"MERGED","created":"2019-11-05 11:01:20.000000000","updated":"2019-12-06 17:14:18.000000000","submitted":"2019-12-06 17:11:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692981-1575652300611-0f63802f","meta_rev_id":"7178425798e74f1cd8ed916bde282c0449a4b834","_number":692981,"virtual_id_number":692981,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 17:11:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-06 15:01:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-11-05 11:06:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-12-06 15:01:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 11:06:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-06 15:01:52.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-06 17:11:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77b8b8dc3aec5aa3db80d084e60b851014154485","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-05 11:01:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e4c2b82058d53edfca8144b6b7faab079db42676","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 11:06:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"743c2a4ab042d52933a87dcde5fbc7ab312fec30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 12:57:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/930c7f7379e84cb58143cd33eb9e564a : SUCCESS in 4m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9d34155d0e4143a9ba9a99a8f4b4ae0b : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/130070db333e4a6cae8fea5780acfd23 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5970c50cb2745d8a71d11307c56ca2f : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0871ace2aa3b4f29b9c1ea91b123774b : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf58d8af00bb4007bc71488e866adb1f : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ce8601b7485a481083c37b5c9c26da48 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/42e7cb5ab387451a9327d34638ece0f6 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f97b04c46074ef5a31d8653d5a8b7f2 : SUCCESS in 36m 47s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3de9babb92bb4ab2b5e2a1d77a66742c : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/cc8db7d9d811463ba6f4e64187ddc640 : SUCCESS in 1h 50m 52s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e7234bf985e04fca912d662693880e17 : SUCCESS in 22m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/927a3fc306f3439ab09197cb50173940 : SUCCESS in 38m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ac8c82b5dcae5b801de64c22be1936a36e3b2e68","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 15:01:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9890fe03dd248ed32eb9317133f25793ea6f36f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 15:02:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4e100db17755f1fc5547d6c6318de00bc2b46910","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:11:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"efd65db2742a6bd8cf0ab261c960e3976350ec71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:11:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e9e5cc7da5a44e78a8f49d1bbc2a266d : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16bcc02a9c8d401ebf02ab4045b318f9 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c849a6ccfb604d8f846a39fd184b2374 : SUCCESS in 5m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3f90790c52ac48c6a7c4b036d714fdb3 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e633983bcfde4848af2fbdbc55e78867 : SUCCESS in 7m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de1a44ec90e54c97b118534cc9ca15b1 : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1acc8b1c0d5f4405b099cdca8e7c95fd : SUCCESS in 7m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fae03bd1175241929906b6529eef694f : SUCCESS in 40m 05s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8d516d61771444bdb01690be49c107c3 : SUCCESS in 1h 11m 41s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/43a3c9d56d3d44279fe6b9402c9b3588 : SUCCESS in 1h 55m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/80dfc5ef871645eab0a2f03117bfea39 : SUCCESS in 20m 53s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4fa14d0af923466198ea7f50613bc9ca : SUCCESS in 30m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b24fe8a293920fb20961e552133d57f18392e94e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:14:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ab8a41737440adbbba90ee91fde498 : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eed2561ce2d713b5fa5d4f9fee62eb763fdb58f6","revisions":{"eed2561ce2d713b5fa5d4f9fee62eb763fdb58f6":{"kind":"REWORK","_number":1,"created":"2019-11-05 11:01:20.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/81/692981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/81/692981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/692981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/692981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/692981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/81/692981/1"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-11-05 10:59:28.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-11-05 10:59:28.000000000","tz":60},"subject":"Stop gathering facts for host command run","message":"Stop gathering facts for host command run\n\nThis will speed up `kayobe overcloud host command run`.\n\nChange-Id: I02adfa5819bff781e1649b81c1023b55de32ebd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eed2561ce2d713b5fa5d4f9fee62eb763fdb58f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eed2561ce2d713b5fa5d4f9fee62eb763fdb58f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
