)]}'
{"id":"openstack%2Fkayobe~692983","triplet_id":"openstack%2Fkayobe~master~Ia2c7c151aca75aadc1efc406383b3966d77d4b9e","project":"openstack/kayobe","branch":"master","topic":"seed-hypervisor-kolla-venv","hashtags":[],"change_id":"Ia2c7c151aca75aadc1efc406383b3966d77d4b9e","subject":"Stop running kolla-target-venv in seed hypervisor host upgrade","status":"MERGED","created":"2019-11-05 11:29:10.000000000","updated":"2020-01-28 16:18:51.000000000","submitted":"2019-12-06 17:11:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"692983-1575652299692-91269b4c","meta_rev_id":"2bd0ae91c6abd11630975459fca6670a5b37e901","_number":692983,"virtual_id_number":692983,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 17:11:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-06 15:01:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-11-05 11:38:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-12-06 15:01:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-05 11:38:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-06 15:01:02.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-06 17:11:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1aba31a080893eca9410e74f4b82d1abd552a7eb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-05 11:29:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae5c67b5fabcf15ab2799f74b8f6aac434260081","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 11:38:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77a010df3cdbc91e315ed5e81ee454696d112cde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 13:34:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bae32b2be9d4f53b30c8349ef02923c : SUCCESS in 5m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/238bfad6c4a34280a437dfbadcdc88c6 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57464e32e4634f3aa7faa64482232b1c : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95182dd82ccc4603b9a3431737a08276 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/893fc6177a6d42408b919c574032ac85 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5baea8752a44b8e809aa54b4c5fbd39 : SUCCESS in 6m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d834c5a100a1410e83c153b6f7601154 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/97520b00ba784023b6724dc94577266a : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/12995b57c2ca4a3e973cb74c51671165 : SUCCESS in 26m 33s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/20423a3eb1e04aa5a0907c2bf92b0a16 : SUCCESS in 1h 11m 13s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/9e51e33ec32d49229eda621fc376d341 : SUCCESS in 2h 01m 57s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f1c4abb112944c38a142d40066b3038a : SUCCESS in 19m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5d90d79acf7344bbbbf6d663e4eb3667 : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"32a6a28dbeb45a395053b3ecdfa83c59728c9180","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 15:01:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"93311604819465c9cdd7558f197512cec531a89b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 15:01:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b5cf638f4c77894964de07a4bea95497b25a52a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:11:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/53ac8e071ebb49288ad5c7bbf05bb038 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b040c0ce447452fb9e86cd7c956132f : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71251d6789494897850ef4c0c2f6a1ec : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6fb86efc4b7f4d6093f4d2440e7cdced : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07337b6f9fe547818a71a2f3209dd253 : SUCCESS in 9m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/48e3e34f2d1248a2a59862af48bd2190 : SUCCESS in 4m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d0ee070d69349949e9be64d5f1b10fd : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/49ce7750d9bc4a3c975b74eb0765e3b5 : SUCCESS in 30m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bfea13bef5b244759fe22ce05f7601ac : SUCCESS in 1h 02m 05s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4eac69887f06474493f18a02cb19aa63 : SUCCESS in 2h 02m 41s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c977b24b5a894d1ebefd1e6bf416a3f1 : SUCCESS in 21m 25s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d03851879ce948d5b08ca3b603915785 : SUCCESS in 33m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b6f3ebcdfdfc9b387dd53ff41f83e5f4ab22d71c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:11:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"19a8a46dcd97a9ee2cad441e9e69843a6bfa0d44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 17:12:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714dc6ce29994d2299860e3e2a427acf : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1},{"id":"e365f71e8ac4e35b2fc195bab520d4e94da6eafd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:18:42.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 8009d970bc4c7b8651c80cb142a27e30b3164415","accounts_in_message":[],"_revision_number":1},{"id":"ce59755002bd5877f340d3c94fa6a4b6a6d988ee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:18:51.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit d2f7ddf8a86827d3fcf9bada0c6155ea01ac9608","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0b0b921a62f82bbb50bc06dbe05880e57d364028","revisions":{"0b0b921a62f82bbb50bc06dbe05880e57d364028":{"kind":"REWORK","_number":1,"created":"2019-11-05 11:29:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/83/692983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/692983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/692983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/692983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/692983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/692983/1"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-11-05 11:28:37.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-11-05 11:28:37.000000000","tz":60},"subject":"Stop running kolla-target-venv in seed hypervisor host upgrade","message":"Stop running kolla-target-venv in seed hypervisor host upgrade\n\nWe don\u0027t run kolla-target-venv during `seed hypervisor host configure`.\nThis playbook is limited to seed:overcloud anyway.\n\nChange-Id: Ia2c7c151aca75aadc1efc406383b3966d77d4b9e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b0b921a62f82bbb50bc06dbe05880e57d364028"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b0b921a62f82bbb50bc06dbe05880e57d364028"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
