)]}'
{"id":"openstack%2Fkayobe~693030","triplet_id":"openstack%2Fkayobe~master~Ib5fb726600debee5df1f7085ea37b583bfccefcf","project":"openstack/kayobe","branch":"master","topic":"story/2006247","hashtags":[],"change_id":"Ib5fb726600debee5df1f7085ea37b583bfccefcf","subject":"Fix kolla ansible virtualenv creation on the seed","status":"MERGED","created":"2019-11-05 15:20:29.000000000","updated":"2019-11-13 15:52:44.000000000","submitted":"2019-11-13 15:49:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693030-1573660197869-99007b45","meta_rev_id":"7890ed27f86c77bc28832badfeab0a80ae846da8","_number":693030,"virtual_id_number":693030,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"date":"2019-11-05 15:47:32.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-13 15:49:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-11-08 12:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-11-11 13:41:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2019-11-13 12:02:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-13 13:56:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2019-11-11 14:56:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-13 13:56:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 12:44:14.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-11 13:41:41.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-11-11 14:56:38.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-11-13 12:02:11.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-11-13 13:56:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-13 15:49:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6561faa89ffc2ab0fd4d9c3f86f92b2766db54b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 15:20:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96df0f93620cb29dbab53c27ca1e16ba02b9ca1c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-05 15:47:32.000000000","message":"Patch Set 1:\n\nI tried removing the controller from the inventory, and the seed test still passed: https://review.opendev.org/693029.\n\nIt\u0027s possible this playbook is not required, given that this virtualenv is also created in kolla-ansible bootstrap-servers. It might be useful for upgrades, to ensure we have the latest packages installed.","accounts_in_message":[],"_revision_number":1},{"id":"df1914ff8ac823cf34f3d75d955e0b6fd9e2f97f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-05 17:49:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa5f54c5071748418ef117490c3102ff : SUCCESS in 7m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8d8b68962eb846f487e389d9a1a23974 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c81ce229d14a48009ac329fe0b63011c : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d3d62ae6d3e4c198d75a17a112240e6 : SUCCESS in 7m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5145da3f622445aa9e96b423485a0a1d : SUCCESS in 7m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55a25d487d334a3b8801cc8d965dc0d9 : SUCCESS in 9m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ece12e5d459425c843f0946294b55b2 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4652af8af5ae436883eb05d57487058c : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b90e0cd406d5465dbf297eff960c1f80 : SUCCESS in 40m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ebeda70bc37140f9bcfbe8e41ffd4149 : SUCCESS in 1h 38m 57s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/86638baf71fa413fb6590fcfd995f151 : SUCCESS in 2h 26m 45s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2cd234bb187248cabdff3fd43e4a7ad7 : SUCCESS in 33m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/70794813a6ca4e17b95fcf0a1a37c30f : SUCCESS in 51m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5686ed49d9fb8c4495f3641523366ecc6f23fe29","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-11-08 12:44:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"22444a1c8cd4ff9083376509a0f40f74e6060388","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-11-11 13:41:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"afba3dd3bb42a4eb326a07feb2581c0526f38768","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-11-11 14:56:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5634fa4a62c8fe3af957bdd097c2371d0466181e","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-11-13 12:02:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0615d53486fd2835967cb4dbe32f44fbfac8f61b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-13 13:56:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"65468de2d692da08a7afa23231514f5b27697e71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 13:57:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf58c886377d65341a9eee80c41e512c0fe352dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 15:49:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4e7e85fc5e894dc18902b1110d17b00d : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac0d201fb8284478bd38555ec3362250 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f734374b75394e8e91dda36df42d9a9f : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5a50bae67e7240afb670cbf345fcb107 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f2f40cbb2bb4c71ba94d85a78a155c1 : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a5f3e5726eb459bbddd47fcca5ef6e2 : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c5f4c3e67b5f431cb0a8cbe518c78bff : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f87138223eca4cecbb3758505751b56b : SUCCESS in 39m 46s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c2614b4704a146faa1b08071261f4244 : SUCCESS in 1h 10m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/47e601acead14e80ab5a8a14cdb67173 : SUCCESS in 1h 47m 08s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/692fb99949c641fe9048cba735c1cce4 : SUCCESS in 25m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4c041b2268e5475caecbaafb2cc62165 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":1},{"id":"88008fd9bff0ce72e7d01ab494ab9e15c6856399","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 15:49:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"128a4914467018c196bc3dae7e205431d16d6ae7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 15:52:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc69f42718e4953b469f7567bf4c89f : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4f8a5de8e90abcf7084712848f47cbf3262d6bf","revisions":{"e4f8a5de8e90abcf7084712848f47cbf3262d6bf":{"kind":"REWORK","_number":1,"created":"2019-11-05 15:20:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/693030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/693030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/693030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/693030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/693030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/693030/1"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-05 15:09:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-05 15:20:19.000000000","tz":0},"subject":"Fix kolla ansible virtualenv creation on the seed","message":"Fix kolla ansible virtualenv creation on the seed\n\nBackport: Stein, Rocky, Queens\n\nIn https://review.opendev.org/#/c/671502/ kolla ansible virtualenv\ncreation was modified to only happen on hosts in the kolla-ansible\ninventory. However, that change omitted the seed. It passed CI because\nthe seed and overcloud are both in the inventory.\n\nThis change adds the seed group to the inventory_hostname lookup.\n\nChange-Id: Ib5fb726600debee5df1f7085ea37b583bfccefcf\nStory: 2006247\nTask: 37368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f8a5de8e90abcf7084712848f47cbf3262d6bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4f8a5de8e90abcf7084712848f47cbf3262d6bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
