)]}'
{"id":"openstack%2Fkayobe~693515","triplet_id":"openstack%2Fkayobe~master~I4e766424607317e5817d26461f7a1e180e9b2146","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I4e766424607317e5817d26461f7a1e180e9b2146","subject":"Add documentation for Bifrost and IPA configuration","status":"MERGED","created":"2019-11-08 10:35:57.000000000","updated":"2019-12-06 18:04:22.000000000","submitted":"2019-12-06 18:04:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693515-1575655462992-0f63802f","meta_rev_id":"d7dc4d9a7ed275c43315c8894314dd8a18345ff2","_number":693515,"virtual_id_number":693515,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-06 18:04:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-06 14:17:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-25 13:44:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-12-06 14:17:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 13:44:56.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-12-06 14:17:01.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-06 18:04:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c274c641586a34586b1171f7c951d75a519e7d53","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-08 10:35:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9941423cd3478f5fd2b404303b5b6a69282a8480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 10:44:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8e06ab1ee1d4c3cb594d19bcfe5703d : FAILURE in 4m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4a9b8b17a2496284bbe28e4195bf21 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"754159b8b9f6b2213a073e4947552057a5f50baa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-08 10:59:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0eb312911228b57fec84fdc7d4fa043c4ad73b44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 11:06:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/792123b9af814825959b1abc7c6d7bd3 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d46aa4dd164f3f9d45b09a985dc953 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"cd3e3116b35258609da7bf47d87a16bd3198eb3a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-11 17:33:25.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"797ff61b52a3e1c5aae7401b29edc0fbf62d7598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 17:44:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/124f2aa3f96a40389853c2eece796e30 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62360987a3f54ccc94401dd8c0c0fed3 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":3},{"id":"546b822720fa917c390153d099a71e315bd88fe8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-15 11:17:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nExcellent work! Overall looks good, only noticed one small issue.","accounts_in_message":[],"_revision_number":3},{"id":"31eacf48141567044e4a83bb24699e0bc00bf70c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-25 13:42:24.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"222c472a757c6e16f1757e459022ed44b44040b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-25 13:42:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dda88b6121fbee800da0d7836bfb4f3a73f1c1fd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-25 13:44:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"29f93b211ec56a2cc40d071bfa4eb3ad1a89806e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 13:53:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fcdf021a38f458bbcbd123237f7b756 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a229e05b9a244925a3781940ec17bf66 : SUCCESS in 5m 53s","accounts_in_message":[],"_revision_number":4},{"id":"17c323196ad3ca6e0af7ebd20354d9b2b73ad334","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 14:17:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"595fb2eb92914f91e425bd7663a5d175f65a476c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 15:22:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"1fac879af0dfe5237202e2cbbcf69a6d5da2d81d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:04:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"d7dc4d9a7ed275c43315c8894314dd8a18345ff2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:04:22.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36bdb1fd487d40f7ad74a6bac508072a : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14f92071d613426d9f1b499f545c3d7e : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8c01771e29daa1f15d441155062ed0ac8c9f95ef","revisions":{"73536dbe80b45851095a2954ddf44034c54b23ef":{"kind":"REWORK","_number":1,"created":"2019-11-08 10:35:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/693515/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/693515/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/693515/1"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-06 16:38:38.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-08 10:34:36.000000000","tz":0},"subject":"Add documentation for Bifrost and IPA configuration","message":"Add documentation for Bifrost and IPA configuration\n\nDepends-On: https://review.opendev.org/690559/\n\nChange-Id: I4e766424607317e5817d26461f7a1e180e9b2146\nStory: 2004337\nTask: 37378\nTask: 37381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73536dbe80b45851095a2954ddf44034c54b23ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/73536dbe80b45851095a2954ddf44034c54b23ef"}]},"branch":"refs/heads/master"},"45ae0802a218a80aeeda4272957a6669c02e8f7a":{"kind":"REWORK","_number":2,"created":"2019-11-08 10:59:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/693515/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/693515/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/693515/2"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-06 16:38:38.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-08 10:59:38.000000000","tz":0},"subject":"Add documentation for Bifrost and IPA configuration","message":"Add documentation for Bifrost and IPA configuration\n\nDepends-On: https://review.opendev.org/690559/\n\nChange-Id: I4e766424607317e5817d26461f7a1e180e9b2146\nStory: 2004337\nTask: 37378\nTask: 37381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45ae0802a218a80aeeda4272957a6669c02e8f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45ae0802a218a80aeeda4272957a6669c02e8f7a"}]},"branch":"refs/heads/master"},"ab04ccb8a27bef7830e8729f88352810c36e9180":{"kind":"REWORK","_number":3,"created":"2019-11-11 17:33:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/693515/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/693515/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/693515/3"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-06 16:38:38.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-11 17:12:28.000000000","tz":0},"subject":"Add documentation for Bifrost and IPA configuration","message":"Add documentation for Bifrost and IPA configuration\n\nDepends-On: https://review.opendev.org/690559/\n\nChange-Id: I4e766424607317e5817d26461f7a1e180e9b2146\nStory: 2004337\nTask: 37378\nTask: 37381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab04ccb8a27bef7830e8729f88352810c36e9180"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab04ccb8a27bef7830e8729f88352810c36e9180"}]},"branch":"refs/heads/master"},"8c01771e29daa1f15d441155062ed0ac8c9f95ef":{"kind":"REWORK","_number":4,"created":"2019-11-25 13:42:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/693515/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/693515/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/693515/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/693515/4"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-06 16:38:38.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-25 13:42:21.000000000","tz":0},"subject":"Add documentation for Bifrost and IPA configuration","message":"Add documentation for Bifrost and IPA configuration\n\nDepends-On: https://review.opendev.org/690559/\n\nChange-Id: I4e766424607317e5817d26461f7a1e180e9b2146\nStory: 2004337\nTask: 37378\nTask: 37381\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c01771e29daa1f15d441155062ed0ac8c9f95ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8c01771e29daa1f15d441155062ed0ac8c9f95ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
