)]}'
{"id":"openstack%2Fkayobe~693523","triplet_id":"openstack%2Fkayobe~master~I0782fb095cc074fd586d87aa338cefb72b1b70e2","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I0782fb095cc074fd586d87aa338cefb72b1b70e2","subject":"Remove unused IPA config variables","status":"MERGED","created":"2019-11-08 11:10:22.000000000","updated":"2019-11-25 20:08:48.000000000","submitted":"2019-11-25 20:07:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693523-1574712436719-a4121484","meta_rev_id":"8963ff8a1e99b5f88effb194bb45e509ee4b12f2","_number":693523,"virtual_id_number":693523,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 20:07:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"all":[{"value":1,"date":"2019-11-11 13:41:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2019-11-13 15:40:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":1,"date":"2019-11-13 12:02:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-11-25 13:39:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-08 11:18:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2019-11-11 14:56:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":1,"date":"2019-11-25 13:39:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-08 11:18:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-11 13:41:16.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-11-11 14:56:11.000000000","updated_by":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"reviewer":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"state":"REVIEWER"},{"updated":"2019-11-13 12:02:42.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-11-13 15:40:45.000000000","updated_by":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"reviewer":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"state":"REVIEWER"},{"updated":"2019-11-25 20:07:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"65291485972ed083893e61a183c1d7776bf1f819","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-08 11:10:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9b7b6020df3793fbcdd374861163ec9c7e28259b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-08 11:18:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe63fe1f8c53cf6dbc7d743f2163cf2d635e747a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-08 12:54:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fb780d1436d4f138efb4ae825aaabaa : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8f4412c0d9e240f0a5489e4ddba9e05c : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ebd5c6458464815a47bf38dc55afbb8 : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/768764c569b54936a52026514cc323f2 : SUCCESS in 3m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1952328c54c744ee83fb4ac9367740fe : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b7bf83c84c4ea1ae6661c4e1d3222e : SUCCESS in 5m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e5342cf8a81144a5916c16ece49b5be4 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3179e511f61e4286a0678e748a104dce : SUCCESS in 8m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65bb2cb43cd6435d92c33da013f281e3 : SUCCESS in 40m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/936e09fb696541ddb76667d8e716a7b4 : SUCCESS in 1h 13m 57s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/388ceffb40c34f2884d99f11a606f4b1 : SUCCESS in 1h 43m 03s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3fcad950562f4325868c5c5842e77ef3 : SUCCESS in 20m 43s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5dc42f324e954073b80731b84a2be474 : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":1},{"id":"487720fadc434f636b67ba175b42d1367c1451a7","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-11-11 13:41:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7217895ca621326c7e1e0623fa9be811d94a7c90","author":{"_account_id":28543,"name":"shupeng","email":"15050873171@163.com","username":"Vieri"},"date":"2019-11-11 14:56:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"783f6e23b84992d1d4f221790871d2fa0f657241","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-11-13 12:02:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"29415aaa53ad524fe836c46601d55fe0fbdb4c02","author":{"_account_id":26297,"name":"pippo","email":"1211099771@qq.com","username":"huang.zhiping"},"date":"2019-11-13 15:40:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c220d8875286b25e0bcc124f0acd7a8061db349a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-25 13:39:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nTrivials","accounts_in_message":[],"_revision_number":1},{"id":"cd70611c8df938d76c45fdf8df641a64494f4bff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 13:39:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4449d616549c45c8f4595a9dd1eb9e93eb8206fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:21:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88f68f7a5c2a4a1bb01adaaee3085149 : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64874fe40c78410990a4ce5b0f2e0be9 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18c7f447b80b4ef9b5cd37b3acabbce7 : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04fdac7086ba4aaaa17435d16f774fd1 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83fc289ccef74a9580cdb16b80bb740d : SUCCESS in 5m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c714f59e522c4676b6936cab8eefd350 : SUCCESS in 4m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6caf1b870083470795ba814434265555 : SUCCESS in 7m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5fa443670b054cf980f79c6aef49df5d : SUCCESS in 38m 23s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f38d37714b714b60b71dae6343640c18 : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/05d94405e1fa469f81847cac39476839 : SUCCESS in 1h 40m 50s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/89d4d3c4f0fa403fb8eb915770494d36 : FAILURE in 20m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/6f15ad65b0824400a3be6a62650cbde3 : FAILURE in 20m 31s","accounts_in_message":[],"_revision_number":1},{"id":"c6a9f3ef0faab25fe35b70feb7b2ad63eb8402cf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-25 16:31:39.000000000","message":"Patch Set 1:\n\nrecheck\n\nmirrors","accounts_in_message":[],"_revision_number":1},{"id":"9ed122ad66b07e8f7f7c379a949008cbe849654a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:14:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/739601b99d7149ba94e67182b0b49345 : SUCCESS in 3m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9df01ba1f1bf4998b5e307621d1a84e5 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c41bf83129a44009158b4437a0057bc : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebd81afd3e534bc09349d9829bda44c7 : SUCCESS in 3m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4cfc58ca34ae4c37b48972e1354f7071 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/277647d0ca7d4c9ba242df727cf7e433 : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a582bade59f43058eebc52a24837a26 : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1a48edeaa5c44e2acd09ae68acf50fe : SUCCESS in 7m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e84c965657a1451091882ca58c1c1b50 : SUCCESS in 29m 19s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/81d81e9809524fffbf753e6482a9a15f : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/08218f27e5d548d2b9d2994a0687af50 : SUCCESS in 1h 41m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/be73106763d0404b9342414c0e147685 : SUCCESS in 19m 25s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/36118fe95adf455baad329a77a360725 : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":1},{"id":"2cda7697c1971bc9a0270055363d8f017727d750","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 18:14:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9b19cbfd4be77227cad2c308de344cfb1d68445","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 20:07:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/649f58662388472083f1c0b74b2821e3 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27430a291aa4409c89d4244428bef11d : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b72a49844bf041ce899fd4bf31eeaf0c : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59000c3c7cd54264aaf7975a52d0742e : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/748d9a20e4e442b0b2fd2163920c259c : SUCCESS in 7m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d3d545939c45416f85ff684d75aa309d : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f6568663926459bb185adf8f243dcc5 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d8022255766f483880511210cf647377 : SUCCESS in 35m 01s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7898d390b0bf440eb3d4c16d120fc2cc : SUCCESS in 1h 07m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c7ce34aff3d640dba8950ddb0221f027 : SUCCESS in 1h 50m 30s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/289e7b6e2f394770b5480193f96dbac6 : SUCCESS in 24m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/52c4dfea4373406c80e3682f62407ad4 : SUCCESS in 35m 13s","accounts_in_message":[],"_revision_number":1},{"id":"effc68f2a87caca51afded59565f09b7b650d5ad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 20:07:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8963ff8a1e99b5f88effb194bb45e509ee4b12f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 20:08:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede01320b3ff48ecbd73a9dca9a49786 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"492605b64db0aa74caafbe50061bc13e9b9ff6a0","revisions":{"492605b64db0aa74caafbe50061bc13e9b9ff6a0":{"kind":"REWORK","_number":1,"created":"2019-11-08 11:10:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/693523/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/693523/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/693523/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/693523/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/693523/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/693523/1"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-08 11:09:19.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-08 11:10:10.000000000","tz":0},"subject":"Remove unused IPA config variables","message":"Remove unused IPA config variables\n\nThese are not referenced anywhere in the code.\n\nChange-Id: I0782fb095cc074fd586d87aa338cefb72b1b70e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/492605b64db0aa74caafbe50061bc13e9b9ff6a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/492605b64db0aa74caafbe50061bc13e9b9ff6a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
