)]}'
{"id":"openstack%2Fkayobe~693686","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I0d3c7865eab71e0020bdb4646550d8ae27e31458","project":"openstack/kayobe","branch":"stable/stein","topic":"chrony-fix-stable/stein","hashtags":[],"change_id":"I0d3c7865eab71e0020bdb4646550d8ae27e31458","subject":"Only stop chrony on nodes in kolla inventory","status":"MERGED","created":"2019-11-11 12:18:17.000000000","updated":"2019-11-11 15:58:24.000000000","submitted":"2019-11-11 15:57:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693686-1573487823515-dfbf8c2d","meta_rev_id":"1f2595c6aac6a22d0b2d07b98f33a258d7bd6947","_number":693686,"virtual_id_number":693686,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-11 15:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-11-11 14:22:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-11 13:29:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-11-11 14:22:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-11-11 12:18:17.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 12:18:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-11 13:29:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-11 15:57:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4feee15095a79565eb6928118633c628b5a85225","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-11 12:18:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4e7788dca1353bf477e1c144f2c15a271f79afd6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-11 13:29:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"427b9e7bf63896483a1c6c0b6d7c32379cd80250","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 14:12:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c154a67199a34c808b7838c3447381f9 : SUCCESS in 5m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b57043fb9b6b41c68ca4417b326b0b7f : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc7ba5ab1db14c64b2326f2968b5ac6d : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87014e90ae344792a651e54b4721797c : SUCCESS in 3m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e8859c6dd68a41578dd2edb58e1ebaec : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1be55f68a48f44aeb697da26b347af52 : SUCCESS in 4m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/15f10fff16c5403a85b6ab0fa348791a : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f1e82e934e84f5faf3b7f0832cfd74b : SUCCESS in 8m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c7d4105906294380998d128371fd5018 : SUCCESS in 37m 30s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ac7ccc2b14284a3ba05ebaf1bfbdad50 : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c43a24406e2a4cf1b2d4e6a10e43e3f5 : SUCCESS in 1h 43m 49s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ca59b66e4cb14c249fd363116ea7f66e : SUCCESS in 22m 02s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/54a0f29ed34349f1b199b86feebe9aac : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":1},{"id":"6a95bca49057ef4536ccd662d8d23acae90be41b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-11 14:22:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8b1f3293fe26b3e58a15cfe15b635b8087a5fd90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 14:22:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0083a39cba600114bef58d9ebcd36fd876b7084c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:57:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4c4055a5628ffd2769a3fa8157daf296fb2b20e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:57:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ccc51fe310234bf8aa51492f7b5931d1 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdd64dac811c45e19e584a4fc7c98a4e : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/794322e40e7d4823ab45c755d1979bf1 : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/97a9cb42177c4c669781ae11e161b53b : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c82e1f0e928b41819fc85a24850b0cc6 : SUCCESS in 3m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec750372eff44f9bbdc3fac7fc52007d : SUCCESS in 3m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8e4442e322ce4b01b22ad977aa07b6af : SUCCESS in 7m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4f94b8d8c84d4ccb928abe1f75a03362 : SUCCESS in 27m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/027468248b7b4e518e06edd244b5b64c : SUCCESS in 58m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/594ea6258dde430a9613d431ceaf2d25 : SUCCESS in 1h 33m 36s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/93e45da50ef345059ba54d89643530e0 : SUCCESS in 17m 07s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5c200aba2b6b4e1a926694b7857e99c0 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b772cb989feec57a0a32e09340e53539b180c08e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 15:58:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6dee8b85214419f940e0c24c27b29e6 : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5137651e91e23f53bf97bde72d3cea3e4993f3b","revisions":{"b5137651e91e23f53bf97bde72d3cea3e4993f3b":{"kind":"REWORK","_number":1,"created":"2019-11-11 12:18:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/86/693686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/693686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/693686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/693686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/693686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/693686/1"}}},"commit":{"parents":[{"commit":"fcdccbc4b24fed76abefa51eb442a240300c75c4","subject":"Merge \"Fix baremetal serial console autoenable\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fcdccbc4b24fed76abefa51eb442a240300c75c4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-11-08 13:52:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-11 12:18:17.000000000","tz":0},"subject":"Only stop chrony on nodes in kolla inventory","message":"Only stop chrony on nodes in kolla inventory\n\nIf docker is not installed this task will currently fail.\n\nChange-Id: I0d3c7865eab71e0020bdb4646550d8ae27e31458\nStory: 2006247\nTask: 37391\n(cherry picked from commit 114b92905ad6842c9d26f0976307cdb63e6174f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5137651e91e23f53bf97bde72d3cea3e4993f3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5137651e91e23f53bf97bde72d3cea3e4993f3b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
