)]}'
{"id":"openstack%2Fkayobe~693687","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I0d3c7865eab71e0020bdb4646550d8ae27e31458","project":"openstack/kayobe","branch":"stable/rocky","topic":"chrony-fix-stable/rocky","hashtags":[],"change_id":"I0d3c7865eab71e0020bdb4646550d8ae27e31458","subject":"Only stop chrony on nodes in kolla inventory","status":"MERGED","created":"2019-11-11 12:18:31.000000000","updated":"2019-11-11 16:19:00.000000000","submitted":"2019-11-11 16:19:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693687-1573489140557-2b47506d","meta_rev_id":"414950bdfb09b59e396bf9cd7082cbed221f94e6","_number":693687,"virtual_id_number":693687,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-11 16:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-11-11 14:23:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-11 13:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-11-11 14:23:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2019-11-11 12:18:31.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-11 12:18:31.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-11 13:29:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-11 16:19:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3456db50c2ffd76618cf15c64256369962040488","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-11 12:18:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cba5dbdca79e39db7bf0928027818d4539b7023e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-11 13:29:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"bc3618dfb0a0310567d4dd9c561f7238407008bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 13:51:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e131a8096317499aa8824c1663f78fdf : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/301c6ddfcae945dc8ee2c8c3bbe5a5b8 : SUCCESS in 3m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64bd833e80274118a1085c6dd0a23b8c : SUCCESS in 5m 05s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/61da3f1ba8684fe7b90c1342b84b64a7 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b96c6ed576b649f99131e945f93f4933 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/896e2dfba79b4848a3fc6c9d4f1a8f0c : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/708b6249fb5c4c44a76a50ebe369c4c7 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23cbb66cb6824512b81feb75475b9e1b : SUCCESS in 7m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aff0a5d6bac34b70a31cc368d86fbf52 : SUCCESS in 26m 37s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e1f99e9702dd4b47a6b8c723eae7d052 : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/fd4fe862823f486085dc05a83793a969 : SUCCESS in 1h 30m 35s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/9486893c84904ac2a68e88539716517a : SUCCESS in 19m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4330b8250ce64cb1a02a7b5280e87028 : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":1},{"id":"fafa9b20f04151a62e1faab7e152333fa95e5f8d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-11 14:23:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"baab7dca9af5d0e2b5fd5f78797548b0aff1aa1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 14:23:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"263bb822b5a65321746088d1ea852a6c02ef1726","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:19:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"afd0e0a249d0bbf3fdb7c896aaf013f3b7bd22fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 16:19:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/2c7999d66a2b4d9c890bb5d99ed089c0 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84b7f9eb84514953bc14db0f070dd690 : SUCCESS in 3m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/602865b8e95d4f6a93e10d21b97087a4 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae68d1a79684a60965ff80d1633a130 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c196271e51a42a1879fb2aac5bfa244 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e0233559e9e649cea4b03a414a723d52 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9fba7d71160441f591903fce56f59d89 : SUCCESS in 7m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fc3dfb16e65142ef8a37800dbaeb95b6 : SUCCESS in 35m 29s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/10f31b500ee04d4893f58b5ac24a691a : SUCCESS in 1h 04m 32s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/1732b6bfbd344bfaa84c0fa198f76cfb : SUCCESS in 1h 51m 08s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/720da104a1cb48458b6421e904d35f28 : SUCCESS in 22m 12s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d8554e1ececf4276801171fb7ca413e6 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"735b66db8523566c35819f8d795f65c2e9cdab33","revisions":{"735b66db8523566c35819f8d795f65c2e9cdab33":{"kind":"REWORK","_number":1,"created":"2019-11-11 12:18:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/87/693687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/87/693687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/693687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/693687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/87/693687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/87/693687/1"}}},"commit":{"parents":[{"commit":"5e1bcb9ae9a58ca5d58772277d1bdc00a2dee3b3","subject":"Merge \"Update packages in virtualenvs\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5e1bcb9ae9a58ca5d58772277d1bdc00a2dee3b3"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-11-08 13:52:01.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-11 12:18:31.000000000","tz":0},"subject":"Only stop chrony on nodes in kolla inventory","message":"Only stop chrony on nodes in kolla inventory\n\nIf docker is not installed this task will currently fail.\n\nChange-Id: I0d3c7865eab71e0020bdb4646550d8ae27e31458\nStory: 2006247\nTask: 37391\n(cherry picked from commit 114b92905ad6842c9d26f0976307cdb63e6174f6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/735b66db8523566c35819f8d795f65c2e9cdab33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/735b66db8523566c35819f8d795f65c2e9cdab33"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
