)]}'
{"id":"openstack%2Fkayobe~693701","triplet_id":"openstack%2Fkayobe~master~I97ada3c6fd94a9ab7b5add97e90717f2a56bc8f3","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I97ada3c6fd94a9ab7b5add97e90717f2a56bc8f3","subject":"Improve docs on updating the Ansible control host","status":"MERGED","created":"2019-11-11 14:13:24.000000000","updated":"2019-11-25 14:19:25.000000000","submitted":"2019-11-25 14:17:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693701-1574691421923-1353a74c","meta_rev_id":"209df1849c08a7510883d2ca26a2a088fa26b4b8","_number":693701,"virtual_id_number":693701,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 14:17:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"value":2,"date":"2019-11-25 13:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-25 13:59:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-25 13:59:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 13:36:39.000000000","updated_by":{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},"reviewer":{"_account_id":18368,"name":"Kevin Tibi","email":"kevintibi@hotmail.com","username":"kodokuu"},"state":"REVIEWER"},{"updated":"2019-11-25 13:36:39.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-25 13:59:40.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-11-25 13:59:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-25 14:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"12d21a8bdeaaf30f86e1a8a0bdbe890615dd1a87","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-11 14:13:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e61f2b8000b2da6767838887af8f78ecb7a34d76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-11 14:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/384e1a15db984e6ba9227e2446de1136 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbeb5a66b9c14aaab743b32ff6d7aa23 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":1},{"id":"a610caa2da14be2ce01c1ba6f980a4d892e23ea9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-25 13:51:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bb64e2928679d0f0a785e4612bd3d95ad1dc07c8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-25 13:57:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3a1ac9beee2e23eeee8b9286ceb0a7b6e88664c1","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-25 13:59:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e2997d62502d6a8a992f3a3a45d84da3db71bb6c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-25 13:59:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6f9080d1e4eae940785673d3633599e2f849f7cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 14:05:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4ca4567dd134d39b9d3eb78b3583ceb : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8f34b6cdd64b53aaef07668760815e : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"867ab07800734c3d6a198079e5ce58a7424e7b29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 14:05:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2ace3fac267253bb42cfc162dda9821a206f6eab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 14:17:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0eb2029d7dd54032045df2d9f293d2059c6beed7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 14:17:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76b95cd5ec0b46ed9a3ee519eb5ad98d : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e67e82e55f34cfc8dc36d88cc020538 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":2},{"id":"209df1849c08a7510883d2ca26a2a088fa26b4b8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 14:19:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c86f98f43744d8bd3573c1a65d1336 : SUCCESS in 1m 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"410466bb5ced328540724fda70bd14fe0197b1a9","revisions":{"454b726efc5801abf556b80e29a3dbcd2f56f9e9":{"kind":"REWORK","_number":1,"created":"2019-11-11 14:13:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/693701/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/693701/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/693701/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/693701/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/693701/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/693701/1"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-11 14:11:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-11 14:11:04.000000000","tz":0},"subject":"Improve docs on updating the Ansible control host","message":"Improve docs on updating the Ansible control host\n\nAlso reorder upgrade documentation to make it clear that configuration\nmigration is something that should be done as a preparatory task.\n\nChange-Id: I97ada3c6fd94a9ab7b5add97e90717f2a56bc8f3\nStory: 2006677\nTask: 36953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/454b726efc5801abf556b80e29a3dbcd2f56f9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/454b726efc5801abf556b80e29a3dbcd2f56f9e9"}]},"branch":"refs/heads/master"},"410466bb5ced328540724fda70bd14fe0197b1a9":{"kind":"REWORK","_number":2,"created":"2019-11-25 13:57:28.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/693701/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/693701/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/693701/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/693701/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/693701/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/693701/2"}}},"commit":{"parents":[{"commit":"a02c65832c13d204d2601ae6560e0863fc4905ef","subject":"Merge \"Fix baremetal serial console autoenable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a02c65832c13d204d2601ae6560e0863fc4905ef"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-11 14:11:04.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-25 13:57:23.000000000","tz":0},"subject":"Improve docs on updating the Ansible control host","message":"Improve docs on updating the Ansible control host\n\nAlso reorder upgrade documentation to make it clear that configuration\nmigration is something that should be done as a preparatory task.\n\nChange-Id: I97ada3c6fd94a9ab7b5add97e90717f2a56bc8f3\nStory: 2006677\nTask: 36953\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/410466bb5ced328540724fda70bd14fe0197b1a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/410466bb5ced328540724fda70bd14fe0197b1a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
