)]}'
{"id":"openstack%2Fkayobe~694220","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ib5fb726600debee5df1f7085ea37b583bfccefcf","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Ib5fb726600debee5df1f7085ea37b583bfccefcf","subject":"Fix kolla ansible virtualenv creation on the seed","status":"MERGED","created":"2019-11-14 09:52:47.000000000","updated":"2019-11-14 14:40:24.000000000","submitted":"2019-11-14 14:37:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694220-1573742229749-bdb6f180","meta_rev_id":"d6e6b2520d70aa3ac79f4dd9d1cb25683172f560","_number":694220,"virtual_id_number":694220,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-14 14:37:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2019-11-14 11:01:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-14 12:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-14 12:32:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 12:32:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-14 14:37:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13823154314dff1d54c0894bf1b31f93c37f4b1f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 09:52:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"861eb87f2293818bc650507ff757fc51d1f11236","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-14 11:01:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2f755067686259a20fe27a5ecbadd175286b7c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 11:36:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60391b6b9acb4c729de5cfb621a6a9b9 : SUCCESS in 5m 22s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/81e2f589d869434faa667a9448e7a6b7 : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9457fc95473946c99a5ad878bbdb52d2 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/604b1be0baea4525aada9b669f6607dd : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/83ac969cc9fc476ba75e3e3b988fbf15 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fae444baebd4d859fea1fa0f53940a6 : SUCCESS in 5m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5a4fd1a16aec40d7bddd876b9008b3d0 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d48a9abfd30a457c9e96b51712579bdf : SUCCESS in 8m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/193ddc4ee2a74d06aacd59496d1f8054 : SUCCESS in 37m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8bbd85183a85456fba12202312bb051c : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/4c51de5d09c44f7f8f6431470d5a64b5 : SUCCESS in 1h 41m 56s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e38bddc3496441bb900d21f1b868d64a : SUCCESS in 20m 39s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/1148a4b789b94ef083f41b9a093095b7 : SUCCESS in 41m 39s","accounts_in_message":[],"_revision_number":1},{"id":"2a6ccf7c3864fc45c9aa1b73e98b9cb447370b5f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-14 12:32:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7cf2af168cca610ea2a81c83a551ab7aeb794621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 12:32:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d1572070ac01d70f42c327d71dad8aec933bb86d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:37:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9e5fa7fd005e414b8ace51130f456f19 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b58f87564f61427cbc63d3a3e1fe8c8f : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa4b04f5145b4975a307edc236592f3c : SUCCESS in 3m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0e4d1330a07d4a5aa7f514be5058ef5b : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faeaf33e025a4995a4e357477989836c : SUCCESS in 9m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/592c25821f494eb7a41d42771020b829 : SUCCESS in 5m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4392fca8cab14ec4803a163bfbf0170b : SUCCESS in 8m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0120ec21f310402aa73aa61659b7cc14 : SUCCESS in 38m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0a974a6b64fd469c9be3c8e37db95e4b : SUCCESS in 59m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/88702c72233c40b0b2c819a8efee943c : SUCCESS in 1h 34m 57s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f31c7fe0fdda4770901185c8d4362134 : SUCCESS in 28m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/551dd7457f9d44debe971efe3fa42ca4 : SUCCESS in 36m 43s","accounts_in_message":[],"_revision_number":1},{"id":"36f12bda854bb1259393da6575b9746658667a57","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:37:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6e6b2520d70aa3ac79f4dd9d1cb25683172f560","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:40:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c64de4be006e4e6a9e3d240686ca105a : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"82522469c977a5626072a077bd691450305e9659","revisions":{"82522469c977a5626072a077bd691450305e9659":{"kind":"REWORK","_number":1,"created":"2019-11-14 09:52:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/694220/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/694220/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/694220/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/694220/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/694220/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/694220/1"}}},"commit":{"parents":[{"commit":"f40b64ebd82b5179a1794a80874060f23fdcb813","subject":"Only set up the kolla ansible venv on hosts in the kolla ansible inventory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f40b64ebd82b5179a1794a80874060f23fdcb813"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-05 15:09:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-14 09:52:39.000000000","tz":0},"subject":"Fix kolla ansible virtualenv creation on the seed","message":"Fix kolla ansible virtualenv creation on the seed\n\nBackport: Stein, Rocky, Queens\n\nIn https://review.opendev.org/#/c/671502/ kolla ansible virtualenv\ncreation was modified to only happen on hosts in the kolla-ansible\ninventory. However, that change omitted the seed. It passed CI because\nthe seed and overcloud are both in the inventory.\n\nThis change adds the seed group to the inventory_hostname lookup.\n\nChange-Id: Ib5fb726600debee5df1f7085ea37b583bfccefcf\nStory: 2006247\nTask: 37368\n(cherry picked from commit e4f8a5de8e90abcf7084712848f47cbf3262d6bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82522469c977a5626072a077bd691450305e9659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/82522469c977a5626072a077bd691450305e9659"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
