)]}'
{"id":"openstack%2Fkayobe~694616","triplet_id":"openstack%2Fkayobe~master~I6006e14df27a180f1438f9eb01b9ebefdddaff72","project":"openstack/kayobe","branch":"master","topic":"stable/train","hashtags":[],"change_id":"I6006e14df27a180f1438f9eb01b9ebefdddaff72","subject":"Use OpenStack Train release","status":"MERGED","created":"2019-11-15 18:05:18.000000000","updated":"2020-01-30 08:34:40.000000000","submitted":"2019-12-12 15:02:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694616-1576162954693-9875c5dd","meta_rev_id":"57e25cbb598be5a14f24101049dea0d1f9e56a16","_number":694616,"virtual_id_number":694616,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 15:02:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2019-12-12 11:00:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-12 10:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-12-12 11:00:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-01-30 08:34:40.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-19 09:54:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-12-12 10:58:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-12-12 11:00:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-12 15:02:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8065018ed6e11eed74c15e1e80e12a484043fce","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-11-15 18:05:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d38114205a2f793028f2104e4384b9f93683e62e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 20:04:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/068d64394ce84371b7d72babbdc97db8 : SUCCESS in 3m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/172f6450e3974f0080e948eb88b74e5e : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6654305ebe5e4ac4a402b0e36a5e3a5e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f554551a309842eabc18524109c377f4 : SUCCESS in 3m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/89c36e957d004074a129b03109c7de13 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ea80807aa5b4941bd8f35cf2ddfea4e : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8982b866b7446698e088ac72903e07b : SUCCESS in 4m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be46caed625a4a1babc6e6edaeaee99c : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9de888ded324e44955318b47d1a38aa : SUCCESS in 8m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/59202e41fbfb410cbc55926c3599c5b4 : SUCCESS in 38m 18s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2e83fbf8bc4c47a18316b52155e2288b : SUCCESS in 1h 14m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d213acd944d3436c94d8374544d5ac2d : SUCCESS in 1h 54m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b5aa1c23e09848858e0f39053836a041 : SUCCESS in 21m 28s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/85ab9730c2a14673bae06cad58e8ee1e : SUCCESS in 39m 16s","accounts_in_message":[],"_revision_number":1},{"id":"20e8acbac56ce8b4fbc72698380a33ab7142c821","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-18 09:46:12.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"97440ea087938b0e57548ae051755d68954722f8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-18 13:28:34.000000000","message":"Patch Set 1:\n\nLooks fine other than Pierre\u0027s comment.","accounts_in_message":[],"_revision_number":1},{"id":"4c5b54d69605a85e05e6a9cacc28220512939d82","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-11-19 09:54:44.000000000","message":"Patch Set 1:\n\nWhy multiply the same deps in multiple tox envs? doing it under [testenv] should be enough. If not - referring to it as deps \u003d {[testenv]deps} is a bit better idea (when you need to add something on top of parent deps).","accounts_in_message":[],"_revision_number":1},{"id":"8e2e071c927d0a1b11cf4a9590988a4611ed3366","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-19 09:58:01.000000000","message":"Patch Set 1:\n\n\u003e Why multiply the same deps in multiple tox envs? doing it under\n \u003e [testenv] should be enough. If not - referring to it as deps \u003d\n \u003e {[testenv]deps} is a bit better idea (when you need to add\n \u003e something on top of parent deps).\n\nThis is just what comes out of the cookiecutter. We could improve it, but let\u0027s do it separately as we\u0027ll need to revert this patch.","accounts_in_message":[],"_revision_number":1},{"id":"bdd7a36600d9849ea5d5de7ed2fa8c2e430803d5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-06 13:17:07.000000000","message":"Patch Set 1:\n\n@Will Can you please update this patch?","accounts_in_message":[],"_revision_number":1},{"id":"defe4e777e1b49ee293e53a48f9534848fa2be69","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 13:35:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"05c0d96f455348a9821431cd4e98463990357cc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 15:29:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94e50935efbc4638a4b86871f3fcecf6 : SUCCESS in 4m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f12668a2e7c6412db5e1eaffa552f95e : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/961a4a3a448b4f589873b98bcd35ccb3 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3e6cce1b9fc64f9185b4e22b28bda23c : SUCCESS in 3m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04d7aa4f7a694bfc8e97eca82842e34b : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33ec3834ce454ff6a02a00d9de0f81e1 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b77c63114ab1478aac7ba5048bdc6921 : SUCCESS in 2m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa4cf5f6339849489982365367f1c6be : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b5620fc010a641c1998511d1e1a67a20 : SUCCESS in 7m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86e885d1326d44c79c588c4e474793e3 : SUCCESS in 38m 47s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f72664122fbe44e88213f712b8053d98 : SUCCESS in 1h 06m 51s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/449c9b88853040908c271dfa2a9eab15 : SUCCESS in 1h 48m 42s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/06779b91eb404efb82449f4448b4c9d7 : SUCCESS in 19m 58s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5ef2cd05c520427b98dd70d7a223d445 : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":2},{"id":"37a0140aea7f52e6b1b0bfcb6063ac7828b10a47","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 15:32:51.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"8cd0ab4cc3fa233ff4819542309020c15e5a4a0d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 15:33:43.000000000","message":"Patch Set 2:\n\nGot another link to update once: https://review.opendev.org/#/c/687499/ lands.","accounts_in_message":[],"_revision_number":2},{"id":"055a17c4e58886ba3dd193cb599f3fd0e7482667","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-06 16:54:51.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"2c3a11c3e3d71aa73fb000b4b189fa4a9a9b67f1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-09 12:01:23.000000000","message":"Patch Set 2:\n\n@Will https://review.opendev.org/#/c/687499/ has landed","accounts_in_message":[],"_revision_number":2},{"id":"c510fb3a87b73f2f2aa21bcdd242e43bbaa18737","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-11 16:11:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c812afe02c3fb428d73a8cec05294ce4743a3c58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 18:02:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e6e66d367e540d19c872b294c6a5e48 : SUCCESS in 5m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/381a41da888542779ab4bcb74dbf87fe : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40182e06f2fd4b33bcbde219b6660949 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83e15e12cbf8444794fc0cbb3798f8cb : SUCCESS in 3m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/69219ae757244e96aba1c387b3669eb0 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49720ae021a74451a6c555beaf450e91 : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7889a868acc649078a9cae0739d02353 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c5b0818531bb45f3addb8c8c0f7cf3a2 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/545d2c94782a4aaab896608c64a78157 : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ea2bb4b2c4e49a798673b247e228faf : SUCCESS in 32m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ece0249fc1e8488c9ecb90e10d2bbcf2 : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d4693fbd6edc45e58833c95b7684d5c6 : SUCCESS in 1h 49m 03s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/f4b108e581b147a2ada3c5d4d911dcac : SUCCESS in 20m 16s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/058099d907564644a85bd760aa997385 : SUCCESS in 40m 52s","accounts_in_message":[],"_revision_number":3},{"id":"b9111e795f8b405f7b09f3f7a12462c62555c8f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-12 10:33:01.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e27b6b9c6145c7dfe9729a7580f98ba47152d5e2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-12 10:56:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"de94ed3da17b0a2298dbec8b99beab2eee255d59","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-12 10:58:03.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"3cd6b38e70e17a44faa09342907931de9b7b4d86","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-12 11:00:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"e12bc38843d5c68ba029341ce1ab6c89b9d5c794","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:43:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9790186787824c5f9c670b436e9ca617 : SUCCESS in 4m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e577d21d1b704b09b07caeeb42f878a9 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d78f16cc9d842a9807807d1876769ab : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0bff0aa22184e16b1081668364ab8e0 : SUCCESS in 6m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35e19e58e7e34c00bc45e2a3eea69490 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fbc1cfc78b34c4897316b5d8d8e6d66 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf27f4578787475da0d7a62c19a491a6 : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3e04e1b98dda4416a49fc497397bfd7c : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/29c76548280444fd9439d1d8b16e21fe : SUCCESS in 12m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cf6175db669f4cffaf450dabd8dabe28 : SUCCESS in 36m 46s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/f5552ff5b382467692eeb7d00907e131 : SUCCESS in 1h 22m 19s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/273c6fdcee7b4f80bd9a61ddd25582d7 : SUCCESS in 1h 43m 37s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c0dce7034bc1410d9bde3cca78657806 : SUCCESS in 22m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d3dac5074c2e49858210063e109ad8f1 : SUCCESS in 42m 07s","accounts_in_message":[],"_revision_number":4},{"id":"43d7ab8ff718cc569b20d547b3ddb42db69ec28f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:43:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6698e41c5ab8e34922600dd8caf0b76634ef6e9e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:02:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"1d7238c36a680aae1de82efc747ff669044c01a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:02:34.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/21181ed3ba8b41719e790e6cd6c68ded : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8fc4375ecbcc41a6aacff65e6815b8af : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b75604366d46358052c57649732a2f : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d3d89516c6384b03bfb050da4a1b8421 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d7a3d1f75e341bbbf5405a91e13276a : SUCCESS in 7m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/58575313b2c6446c885e7c8528b367a3 : SUCCESS in 4m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/895406f9dae9452896ffa5bb99292bea : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/64281ba20bd74200a80c40eaa17b2735 : SUCCESS in 8m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f16af312ce84790a3a2970689e742cd : SUCCESS in 44m 54s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/293dcca22fac46d99792ba623a04df5e : SUCCESS in 1h 19m 35s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a8b36eaa9ae240edbc4e0dba94eedb13 : SUCCESS in 2h 17m 39s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4abda3105be441dcb76039122ec19e66 : SUCCESS in 22m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/94f8c408c7df4dd6aa728efa2737df98 : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":4},{"id":"c4de078936422e6cd649fe70be98f37792d7f1f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:07:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28891e37c1824204b0f2ecb150dacc94 : SUCCESS in 1m 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/347fc8832e21435ab80fb44bae66c9d6 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":4},{"id":"035fbd588838483eb46c98b0f3446df21054f684","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-01-30 08:34:40.000000000","message":"Created a revert of this change as I020b3fc842f8aafef26e2f2425cd511544484baa","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"a93b85ba07113a77a6c0db498a59cef3bca15b94","revisions":{"e6c1ce89a690454b145b175b371127968270cc3f":{"kind":"REWORK","_number":1,"created":"2019-11-15 18:05:18.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/694616/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/694616/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/694616/1"}}},"commit":{"parents":[{"commit":"27fa6b10afce9c7c0a4ba45b8b6b4654051ad5ee","subject":"Merge \"Add reference to zero touch provisioning blog in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27fa6b10afce9c7c0a4ba45b8b6b4654051ad5ee"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-12 14:40:02.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-11-15 18:04:37.000000000","tz":0},"subject":"Use OpenStack Train release","message":"Use OpenStack Train release\n\nSwitch all dependencies to use the OpenStack Train release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/train branch has been cut and RC1 released.\n\nChange-Id: I6006e14df27a180f1438f9eb01b9ebefdddaff72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6c1ce89a690454b145b175b371127968270cc3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6c1ce89a690454b145b175b371127968270cc3f"}]},"branch":"refs/heads/master"},"acc7388f8da1deb5337f050b8659f535b4083230":{"kind":"REWORK","_number":2,"created":"2019-12-06 13:35:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/694616/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/694616/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/694616/2"}}},"commit":{"parents":[{"commit":"27fa6b10afce9c7c0a4ba45b8b6b4654051ad5ee","subject":"Merge \"Add reference to zero touch provisioning blog in docs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27fa6b10afce9c7c0a4ba45b8b6b4654051ad5ee"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-12 14:40:02.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-11-19 12:28:42.000000000","tz":0},"subject":"Use OpenStack Train release","message":"Use OpenStack Train release\n\nSwitch all dependencies to use the OpenStack Train release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/train branch has been cut and RC1 released.\n\nChange-Id: I6006e14df27a180f1438f9eb01b9ebefdddaff72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/acc7388f8da1deb5337f050b8659f535b4083230"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/acc7388f8da1deb5337f050b8659f535b4083230"}]},"branch":"refs/heads/master"},"544cd833a1dc23aef824364881c9cd438482af33":{"kind":"REWORK","_number":3,"created":"2019-12-11 16:11:21.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/16/694616/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/694616/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/694616/3"}}},"commit":{"parents":[{"commit":"074c1c249a56c63ed760348529b61d0e95d5a035","subject":"Merge \"Switch to generic package module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/074c1c249a56c63ed760348529b61d0e95d5a035"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-12 14:40:02.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-12-11 15:53:15.000000000","tz":0},"subject":"Use OpenStack Train release","message":"Use OpenStack Train release\n\nSwitch all dependencies to use the OpenStack Train release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/train branch has been cut and RC1 released.\n\nChange-Id: I6006e14df27a180f1438f9eb01b9ebefdddaff72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/544cd833a1dc23aef824364881c9cd438482af33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/544cd833a1dc23aef824364881c9cd438482af33"}]},"branch":"refs/heads/master"},"a93b85ba07113a77a6c0db498a59cef3bca15b94":{"kind":"REWORK","_number":4,"created":"2019-12-12 10:56:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/694616/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/694616/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/694616/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/694616/4"}}},"commit":{"parents":[{"commit":"9332c211f47a71bd214ad94c5d328078d3ccf199","subject":"Merge \"Add kayobe as openstack project for release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9332c211f47a71bd214ad94c5d328078d3ccf199"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-04-12 14:40:02.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-12 10:55:56.000000000","tz":60},"subject":"Use OpenStack Train release","message":"Use OpenStack Train release\n\nSwitch all dependencies to use the OpenStack Train release.\n\nThis commit should be reverted on the master branch once the\nKayobe stable/train branch has been cut and RC1 released.\n\nChange-Id: I6006e14df27a180f1438f9eb01b9ebefdddaff72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a93b85ba07113a77a6c0db498a59cef3bca15b94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a93b85ba07113a77a6c0db498a59cef3bca15b94"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
