)]}'
{"id":"openstack%2Fkayobe~696100","triplet_id":"openstack%2Fkayobe~master~Ia81c3aba348ab8011d94a5f1584856df1c8361a3","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ia81c3aba348ab8011d94a5f1584856df1c8361a3","subject":"Synchronise kolla-ansible inventory and globals.yml","status":"MERGED","created":"2019-11-26 14:40:48.000000000","updated":"2019-12-09 11:15:36.000000000","submitted":"2019-12-09 11:14:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696100-1575890041561-aff7a9a8","meta_rev_id":"79ea4b7a130c731dcc13239e8295d66ddb1617a4","_number":696100,"virtual_id_number":696100,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 11:14:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-06 17:52:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-12-09 09:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-12-09 09:28:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-12-06 16:53:38.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-06 17:52:41.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-09 09:28:45.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-12-09 11:14:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d577c4dff44e1fee81a8d608bedf3068948a300","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 14:40:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73a7db33d665592682c861301269f18ae41133cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:38:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb5cbd62675048568ecbbf52b1565610 : SUCCESS in 4m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3347b77b1b0f48eaa0dea47c5d4af1e3 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69f45420dd37419f916df637610e3b05 : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7172884ce97647159ec5c0029e38343a : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2b715c6e52a241d38c6e277763abc69e : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84233fccee5f49dfb666b07d8c0cdbe4 : SUCCESS in 5m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/850535b8c4ea4c989d299139f3b64252 : SUCCESS in 4m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f38f798221154154aa661ff399b87ed3 : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6decdfaf6f474bf48afce120c1f76c6d : SUCCESS in 43m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7cebd05979c44781b31a7f908b2692cd : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/40bc1f6131c64c1ba48ab3bd43d76f05 : SUCCESS in 1h 51m 57s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2934c8bb5f504ebcbf910048dfc2f08a : SUCCESS in 31m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e357fb4afc30450daaf7a790b574eb67 : SUCCESS in 38m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9339c3fed717a0493e5db2caafdd691334eafbfc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-06 16:53:38.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"87b2e35f1cae25d5c7550084dcf708b6976e7d7d","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 17:52:41.000000000","message":"Patch Set 1: Code-Review+2\n\nCouldn\u0027t see anything obviously wrong (but hard to check everything)","accounts_in_message":[],"_revision_number":1},{"id":"75e073a1d162fe1d2690233cc6a85302fc5e4127","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-09 09:28:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b7e6265e1bd1b23b53c4789c5c9a5ad51ce0ec5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 09:28:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a51a966e72eb1d244d48a887f18e36ca2e3b20a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 11:14:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/06f8a5f16ccf49d68de121f9366bd162 : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fdbf46937fe4aeab71579fa97b76178 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28bf952e3504b2f937b0288d61d13b5 : SUCCESS in 3m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/682a69d03d7d469a81a361fad9c9ea86 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/573c2b3b77294c8c9314a88f8236ec21 : SUCCESS in 5m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/39856ab4929143dcbcd5a6eabf19c7bc : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/46e4527eeef84140a5e8dfb1e4467f7c : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2dab881386074207a3e8ed71781f866d : SUCCESS in 30m 17s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1e74811148954271bf5968e5c7fa5975 : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5d234e734f8f4157bdbd31f5072dc05b : SUCCESS in 1h 44m 04s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2d103938a07a441ea8a138b1795e8c15 : SUCCESS in 19m 39s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e8e437fc3a7e4fd39b89908d27b71692 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"69a3d5e960a64577cbebfb96d1aa8a0dada346e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 11:14:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"79ea4b7a130c731dcc13239e8295d66ddb1617a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 11:15:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd0eb8fb51544a4ea1ba097159b6ec56 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ecf4985d9cbe5f04489f21a69b5122e2a4244691","revisions":{"ecf4985d9cbe5f04489f21a69b5122e2a4244691":{"kind":"REWORK","_number":1,"created":"2019-11-26 14:40:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/696100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/696100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/696100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/696100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/696100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/696100/1"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 14:23:16.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 14:39:44.000000000","tz":0},"subject":"Synchronise kolla-ansible inventory and globals.yml","message":"Synchronise kolla-ansible inventory and globals.yml\n\nSynchronises the globals.yml template, list of feature flags, and\ninventory templates with the current state of kolla-ansible\u0027s\nstable/train branch.\n\nChange-Id: Ia81c3aba348ab8011d94a5f1584856df1c8361a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecf4985d9cbe5f04489f21a69b5122e2a4244691"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ecf4985d9cbe5f04489f21a69b5122e2a4244691"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
