)]}'
{"id":"openstack%2Fkayobe~696102","triplet_id":"openstack%2Fkayobe~master~Id7b574cbc6bb5195eee16cfc8443075712ce3254","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Id7b574cbc6bb5195eee16cfc8443075712ce3254","subject":"CI: Remove workaround for upgrading from Rocky","status":"MERGED","created":"2019-11-26 14:52:58.000000000","updated":"2019-11-28 18:54:55.000000000","submitted":"2019-11-28 18:53:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696102-1574967229860-2a9069a0","meta_rev_id":"0bcd83ff136a47b65cad9897e146b18af20b2452","_number":696102,"virtual_id_number":696102,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-28 18:53:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-11-28 17:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-28 16:26:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-11-28 17:08:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-28 16:26:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-28 17:08:03.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-28 18:53:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a468878852fd3575d92f700adac66818adafec9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 14:52:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fcf1562ea460870566f8547fb23718c770a74d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:51:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/750b8a7b80dd4ad7ad0ab709f94c60e6 : SUCCESS in 3m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3bbb253cb75a41269345839847a9cba1 : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/138cfbae7f11438389cffb96b0f2c77c : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2a047ad71d0466096012eb596e02ddd : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8e0a6e1c8e4147fdb9d592be660525ce : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0bf840a51bb426c8cede0478e4d0585 : SUCCESS in 10m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9925ac3c6c6d4555a971460629f2ab55 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/757353b6fa7145acb94a9c7063d2784a : SUCCESS in 9m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/da3315370b80416dbff63411894f2f96 : SUCCESS in 36m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/61803513494d471583d5bd8b12bba281 : SUCCESS in 1h 20m 27s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/13b632885b0c4ac5b84fa6e07804fbe8 : SUCCESS in 1h 50m 39s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/734c275d7ccb447fb1d2df696bfb5964 : SUCCESS in 22m 19s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f85b4fcfd5a84e9c8a2e88c11b5d4177 : SUCCESS in 42m 40s","accounts_in_message":[],"_revision_number":1},{"id":"7a15a40a06220ee017351b636bc8a152676d5173","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-28 16:26:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3c097891c1486d056e8c62d2c3e4208dff92b9c2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-11-28 17:08:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dc08f5166d97950232d7510c025acc6a94aaa5eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:08:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a29e13d8ce445f0d47ba3f43a030ce146d0ae6d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 18:53:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"89db810772223c96ec604a75aac984c1a7bad128","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 18:53:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2a097e0c7644780aab31724d516572e : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cd9ef9af0da408eb359ef55e0f34064 : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3bdf3bc4a77493c991df6f1d5ede620 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9358f82344444fd2972e23ab7ba99381 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e81e8e4dc7e4f159a1ddfde65571573 : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b64b258e83441c2940134a7fe4070c6 : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5e67686501b848879042fb25ff9cc6a0 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8eb89d2d20b403d922e5fb76f3ea222 : SUCCESS in 38m 25s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/61ef867f56994f5684edbdc3edbf161a : SUCCESS in 1h 20m 25s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8fcf3243a581476b827b1f9806bf5fa8 : SUCCESS in 1h 42m 59s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/64e1149d948641b785baca4bcc6f18c8 : SUCCESS in 20m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/9811767926e0458e937e5b469a58a54a : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0bcd83ff136a47b65cad9897e146b18af20b2452","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 18:54:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a8e2f9bc64f4caaa4d03bfbc3a4fe6f : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c619579fc125f9d5088210a7832e2603521e797c","revisions":{"c619579fc125f9d5088210a7832e2603521e797c":{"kind":"REWORK","_number":1,"created":"2019-11-26 14:52:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/02/696102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/02/696102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/696102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/696102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/02/696102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/02/696102/1"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 14:51:25.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 14:51:25.000000000","tz":0},"subject":"CI: Remove workaround for upgrading from Rocky","message":"CI: Remove workaround for upgrading from Rocky\n\nWe no longer need to cd to the kayobe repo path, since all tested\nversions of kayobe support pip installation.\n\nChange-Id: Id7b574cbc6bb5195eee16cfc8443075712ce3254\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c619579fc125f9d5088210a7832e2603521e797c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c619579fc125f9d5088210a7832e2603521e797c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
