)]}'
{"id":"openstack%2Fkayobe~696121","triplet_id":"openstack%2Fkayobe~master~I999dfe4b1d082bc88148f54b236644abcbd96a5a","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I999dfe4b1d082bc88148f54b236644abcbd96a5a","subject":"Add command to deploy only containers","status":"MERGED","created":"2019-11-26 16:23:40.000000000","updated":"2019-12-09 11:23:22.000000000","submitted":"2019-12-09 11:21:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696121-1575890498250-758459c9","meta_rev_id":"1ef6e623101045935504866dcf00c41265b6e7e5","_number":696121,"virtual_id_number":696121,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-09 11:21:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-06 17:47:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-12-09 09:38:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-12-09 09:38:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-12-06 16:53:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-06 17:47:06.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-09 09:38:55.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-12-09 11:21:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"02da535f0bc0f72ca793fcfd0959aac88dbadabc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 16:23:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"774d65593563535efd0654c52a6040a262b64f1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 18:47:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6f69b58b0bb342428dcfc9af5a3a8775 : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c568daf91fc64cd781f941041e59c831 : SUCCESS in 6m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/093081d021a243c2bf0d69f026c10f5d : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28e092906d35412292371865f11d7034 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c35bc04007e44b9ea88c49a68c335467 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4b08d6667642407db4a26e1d356694a3 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c75b6635f64a7c9b1bef188b426225 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d5a80274796445d8fe7e631ef5ccfc9 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d2a851e1b20045da94b2cde1aeb04d30 : SUCCESS in 6m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4492f2c8f8924e118821f91b0c791b90 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ecbdb8d866d42288113b0e5670b87ef : SUCCESS in 39m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d5fa5751a0874056963d0a949c7ddbb2 : SUCCESS in 1h 26m 28s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/72acf04f91a74bb697932ff2df56e8d8 : SUCCESS in 2h 06m 57s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7e46b97adfa944869986eca096cea192 : SUCCESS in 25m 28s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/50cabe91feb44ccbbdf04e1b77b14c8a : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":1},{"id":"0d96330105e8d57d6ef9899df46058c48070fdbe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-27 17:05:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"ada68ea394d65279b544950bcceaeaacb5ffbd97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 19:04:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/de081d256a2c42a2b0f868202dbddd68 : SUCCESS in 1m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e53934d7c784d40b21d7992c379f9c4 : SUCCESS in 3m 55s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/906d21901ed54bf5b3c8d22e277e1512 : SUCCESS in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06a79979501a4313b1f51afc5714e13d : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfb9e79482a046dea7a815f57ae46145 : SUCCESS in 3m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6ca39963af61430c8ede0f04e30c1cc7 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1930bdb32b46471ea92b2265de9ba3ce : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d502c9484e974651b0ad23a5420aeb29 : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f3a410dec958482c9cf3746da18605f8 : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/baa7f61bdfce46f5b8f9c83d46a9ba40 : FAILURE in 8m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/616d4a4d19764bf4a2c347451d77097a : SUCCESS in 37m 45s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/10abcd9adf07454bbe757b6a09f2323d : SUCCESS in 1h 13m 42s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7ca780e45dbb4e68ad79194626395cdd : SUCCESS in 1h 55m 18s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6a553d1f251e465f969d24873602a959 : SUCCESS in 27m 49s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3effb54ec6a349b68f8430d8c2d9b32c : SUCCESS in 34m 31s","accounts_in_message":[],"_revision_number":2},{"id":"086ef9774bd43d4d03e6f1ffa51c0bad051d473f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-28 13:31:21.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"084b2bbb9171dd2d036af830d742159d0203ef31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 15:16:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b180d83089564766ac85af52fde9ee67 : SUCCESS in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8bf5645f8e55426e943944fed8816d9f : SUCCESS in 6m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b66c047dff9c4470bdcfe0f078952969 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/143e2890cc5f41d8a003a0f8feb75268 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d7905bb39e49e8b07cab414bcf2a8f : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bec6829c1456479b9dd9e079a98b8a95 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42fe2d95d65d46349fc32cc6d1c3a219 : SUCCESS in 5m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fae658a4f9847ae9bed0b5bd58db0be : SUCCESS in 3m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/982f4732156c44a5896c398e6f2a25b8 : SUCCESS in 7m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fc8caa8f01c1473088ae9947f95a0108 : SUCCESS in 9m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/02c2ecf31e0a4cc5883669ff7a19fa73 : SUCCESS in 31m 12s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/04ac62d05cd34f30b7a20dc47eb61a55 : SUCCESS in 1h 09m 16s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c6dd4aca1d7a40929b4eeb114cec16fe : SUCCESS in 1h 41m 52s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/cb7e294175034aa69e3fafdf4033b3ab : SUCCESS in 27m 41s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/852dd0c5034d49c69bc15a4c49ddff48 : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":3},{"id":"ea321007de24cd20ccc0020bb4a9bd46aee90db1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-06 16:53:30.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"93c390008cac5ffe4cd3b54d5561c8d5e3ffcc89","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-06 17:47:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"18d59f5c2742c6e80725b5c6985cc4c3d574d259","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-09 09:38:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6cd7a023a781f083f0d074ed14ec6d47263941f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 09:39:07.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9ebe1b5aa614597511d34070d691617234052700","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 11:21:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"afd4a6910e2f599c08baf97c12ae7218919f3e08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 11:21:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/976a0d8b54a14266b5807dce0ac350c8 : SUCCESS in 2m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1a767648c21840f79ad72db9356f239c : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba11f95db0a4d6a986069ed805fbabe : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6adc891dc5f344b59b65b2f82fbc7527 : SUCCESS in 3m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be132dbee8b64065a793ca782600b3ab : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91b8ef504a094d8aaf5471492f5c1fe1 : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72c9892ce389482eb142ab4ca38f9586 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c81ad7aa0a642729ae0a0449090d3e0 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/231fd209cac942c2b2282bce7f5ea1a0 : SUCCESS in 8m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2cc3755255884efb9617a04e5ca7221d : SUCCESS in 29m 17s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d9f5619b61dc49b8876f6cae472bf0ee : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/612caea948c843a48277591615945c33 : SUCCESS in 1h 41m 41s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/95e7d690ba114a26be825476732461b5 : SUCCESS in 21m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/79f987763c54443081219b9d1863de16 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":3},{"id":"1ef6e623101045935504866dcf00c41265b6e7e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 11:23:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84166457e5e341028bd0cd29e4dcb5fb : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e92ff45466441dabf9e60214c7082ef : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"66cb51a8bff12614154da6b14abb1165e598c462","revisions":{"18773add6f9de7b11b3c0379c5280b716ef38e45":{"kind":"REWORK","_number":1,"created":"2019-11-26 16:23:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/696121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/696121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/696121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/696121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/696121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/696121/1"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 16:19:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 16:19:50.000000000","tz":0},"subject":"Add command to deploy only containers","message":"Add command to deploy only containers\n\nKolla ansible train adds a new command, kolla-ansible deploy-containers,\nthat will only deploy containers, and skips registration, bootstrapping\nand configuration.\n\nThis change adds a new \u0027kayobe overcloud service deploy containers\u0027\ncommand to make use of the new kolla-ansible command.\n\nChange-Id: I999dfe4b1d082bc88148f54b236644abcbd96a5a\nStory: 2006951\nTask: 37633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18773add6f9de7b11b3c0379c5280b716ef38e45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18773add6f9de7b11b3c0379c5280b716ef38e45"}]},"branch":"refs/heads/master"},"5d7e934d77bcb6a2d833e0080d322af08f2fdf32":{"kind":"REWORK","_number":2,"created":"2019-11-27 17:05:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/696121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/696121/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/696121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/696121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/696121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/696121/2"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 16:19:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:05:08.000000000","tz":0},"subject":"Add command to deploy only containers","message":"Add command to deploy only containers\n\nKolla ansible train adds a new command, kolla-ansible deploy-containers,\nthat will only deploy containers, and skips registration, bootstrapping\nand configuration.\n\nThis change adds a new \u0027kayobe overcloud service deploy containers\u0027\ncommand to make use of the new kolla-ansible command.\n\nChange-Id: I999dfe4b1d082bc88148f54b236644abcbd96a5a\nStory: 2006951\nTask: 37633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d7e934d77bcb6a2d833e0080d322af08f2fdf32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5d7e934d77bcb6a2d833e0080d322af08f2fdf32"}]},"branch":"refs/heads/master"},"66cb51a8bff12614154da6b14abb1165e598c462":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-11-28 13:31:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/696121/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/21/696121/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/696121/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/696121/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/21/696121/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/21/696121/3"}}},"commit":{"parents":[{"commit":"994d7902439789a63cf42caf1c0102349a2b9fb2","subject":"Fix kolla-ansible role tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/994d7902439789a63cf42caf1c0102349a2b9fb2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 16:19:50.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-28 13:31:21.000000000","tz":0},"subject":"Add command to deploy only containers","message":"Add command to deploy only containers\n\nKolla ansible train adds a new command, kolla-ansible deploy-containers,\nthat will only deploy containers, and skips registration, bootstrapping\nand configuration.\n\nThis change adds a new \u0027kayobe overcloud service deploy containers\u0027\ncommand to make use of the new kolla-ansible command.\n\nChange-Id: I999dfe4b1d082bc88148f54b236644abcbd96a5a\nStory: 2006951\nTask: 37633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66cb51a8bff12614154da6b14abb1165e598c462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/66cb51a8bff12614154da6b14abb1165e598c462"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
