)]}'
{"id":"openstack%2Fkayobe~696129","triplet_id":"openstack%2Fkayobe~master~Ib95771f09fd6d5e71a2af471de47f811e1cab88b","project":"openstack/kayobe","branch":"master","topic":"bug/1843043","hashtags":[],"change_id":"Ib95771f09fd6d5e71a2af471de47f811e1cab88b","subject":"Use mariabackup for database backups","status":"MERGED","created":"2019-11-26 16:47:26.000000000","updated":"2020-01-28 16:22:57.000000000","submitted":"2019-12-12 13:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696129-1576156065951-494bf100","meta_rev_id":"13d2525cb2844821cc466af3a02a35d629069d76","_number":696129,"virtual_id_number":696129,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 13:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-12 11:06:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-12 10:30:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-12-12 11:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-01-28 16:22:57.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 10:30:03.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-12-12 11:06:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-12 13:07:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c0be66bcf26f2dec1f2031964b23f32506debe13","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 16:47:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de83b3eafa24dfa276213e4a8f78bade276dae53","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 16:47:48.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"43378dc24b57376accbd67e3ac7e056a4eda47cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-26 16:48:07.000000000","message":"Patch Set 1:\n\nNeeds testing","accounts_in_message":[],"_revision_number":1},{"id":"f6ae05916977b78add27fd407b387c0c9c4767ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 18:34:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96e43e29b97b447aab86e45124b3f34d : SUCCESS in 5m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cc2aa8bee53c4bbd9c379ce08f1458c2 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fa28872a9088457691a747e06a3bceb5 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/891f179d71b64a04868868a95bd64a2e : SUCCESS in 4m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b427f073a41e41e48fea73e36349e0ea : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01923085401745ea870d825a73ed478c : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3020fc8545f34a8c96ce5c31ba2b2402 : SUCCESS in 3m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c41f8a695cab4599809f9e7c224a4617 : SUCCESS in 4m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0fa13981dabb43d0a6ee1b81dabf4c30 : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/14e13d51d0dd4eecb3a71294f732cf2c : SUCCESS in 41m 32s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8559d8723b93435599106a80462efc68 : SUCCESS in 1h 14m 32s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5beaad70a3194536913f51cdd8037618 : SUCCESS in 1h 42m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/72a515e66ca24af685ea881a0740ee02 : SUCCESS in 20m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5dd4da8e2930441db5d8938a847adfea : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"bd33b06fa087aa7e81850b887a8ab87eaeb44a2d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-27 17:04:37.000000000","message":"Patch Set 1: -Workflow\n\nSeems to do the trick.","accounts_in_message":[],"_revision_number":1},{"id":"5d8c9d0783169495341f763482fec10e299f76af","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-11-28 21:42:06.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"abf29f26738f23b653eba4f42b5d0888c7ad3cb7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-04 12:20:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"77a2955f35a505858552c0738ecb61666b2bb46c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-04 12:21:23.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"228612d1b95c3638151f13b962939bd970c274f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-04 14:19:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ae220412c3e4868bceb2ee9a0b430c8 : SUCCESS in 3m 52s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8a2482ca0d2e4a26b9a2bbbdff85d2e1 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dbfa7a5e0aa46f9a9c5b1c782cac660 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2df0dcab50d74c26ba18e9d300c09f6a : SUCCESS in 3m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9122141bf381461fb3744c27658b9364 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad6fae7b06e243ebabeb72764dfd48be : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8d04b8270594fce831b8bc878439e26 : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a09e82323ea8466ea67ade81bcde82d8 : SUCCESS in 5m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f99e60a040f44ce99d9a014cef6bf85e : SUCCESS in 8m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ca52e15ed194c969ec1e16074ec91fa : SUCCESS in 29m 11s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/948b1aa4d27542eda01436f63098dc17 : SUCCESS in 1h 18m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/11c806dfc41646148865149b5fb92995 : SUCCESS in 1h 52m 53s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4854dd278aa14f9dbbf40179de83d2f0 : SUCCESS in 19m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c212aca7988845bdb078d58c7a3460ae : SUCCESS in 44m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b56c78a540122de2c121dcbf157dbdbbc185d487","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-06 16:53:18.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"8ff09b83933002cde811041e970c3f43d26899e9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-11 16:58:40.000000000","message":"Patch Set 2:\n\nneeds a rebase","accounts_in_message":[],"_revision_number":2},{"id":"c85b9dd1d9b1f8269e8124e14d20d976229ac8d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-11 17:17:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7c536017d1e1dc79995349863396324aa7386ec3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:17:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/769d31f6a54b400dbcada45038f4cc52 : SUCCESS in 4m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a09f99acb5504ed8a2a7172aee219c55 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b21d9d79169a4685a7e23e2dc401f0c9 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eee47ee196e2447684a1e4cecf19996a : SUCCESS in 3m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3737f595db704724a65b4f25972f7ab1 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00d7623147c34b10a800d8fd92063189 : SUCCESS in 5m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c1079ce68974af08caf8ca94ac98b99 : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f62e7a9223c477ea9fc15ce77ea6905 : SUCCESS in 3m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d5648b37429e4a17b37d48bc68d293ef : SUCCESS in 8m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4b92913f21da4f389a95e5c5288cd3dd : SUCCESS in 38m 34s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/df103653ff244cb78dc98b358d957322 : SUCCESS in 1h 08m 01s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/28629f10a88b4d9bbad0102c6342cc55 : SUCCESS in 1h 46m 49s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dc6ea4d3b6c94c0d9c288f75dc38480b : SUCCESS in 17m 00s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/6783897963a9474691b1f68c2720875c : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":3},{"id":"8891694f80f49b1c50b1384ca13b9c85c7ddcf00","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-12 10:30:03.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3dbc91dbf2053d27f88b9273d63b7131762e6511","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-12 11:06:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"737a2eb76f8620c2bdf3fc97dc8f7633f80a51e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 11:06:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ff1c8a64b3c3db1ae2ed1f5957cdd2ff088b4efe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:07:45.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/dea6ff388b6d49059ec494104b5d1a79 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15c73fd8c0eb4defa9c05d11afca9647 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c29ff9b33700453a9adf63f74b29eab3 : SUCCESS in 3m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/704632dc7b8a4e79b9fe7877c9edd27f : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9781bdaaf83947b684300f3d018eeef9 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/518a2f0d06394064a7e8202ecb227bf0 : SUCCESS in 4m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a692d9a9921a4b1fb3c59089cc98c3ac : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/80aca67150de4bcca0264b45c1e241b1 : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/617cc527ab2f42dda4da5b220eb084a3 : SUCCESS in 42m 47s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/21745897815d4e9ab2f1f04bdbb8ca4f : SUCCESS in 1h 17m 21s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/62ea2809f5424ff1b107785f0a730459 : SUCCESS in 1h 50m 09s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ad450d6546254108acc24f684fe52222 : SUCCESS in 29m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7ec69cb3ec344ac4b4b1fe2fa90b08ce : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":3},{"id":"bd0c54faf203ad42cc7ecea788f8183f6f8c81c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:07:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9554a74700b2d5fbc9eadd2ab3a5bc4920a1d3fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:10:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad39d069d2ee47b98db81c5a8a8f9672 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20e1d3b9e0f94c29a933c448fbc71d80 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":3},{"id":"850578c6377645ed77781e2de1ccfee1c52378bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:22:57.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 93f1aaab4c08ca8b303bc03500a8b1d4f276dbba","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e93bbd9ed568e98507c792244df58672ab3d84ef","revisions":{"d10fade5a1f38b01f987d5d857a886c173b101d7":{"kind":"REWORK","_number":1,"created":"2019-11-26 16:47:26.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/696129/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/696129/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/696129/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/696129/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/696129/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/696129/1"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 16:24:46.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 16:47:12.000000000","tz":0},"subject":"Use mariabackup for database backups","message":"Use mariabackup for database backups\n\nKolla ansible switched its database backup support to mariabackup from\nxtrabackup due to incompatibilities. See\nhttps://bugs.launchpad.net/kolla/+bug/1843043 for details.\n\nChange-Id: Ib95771f09fd6d5e71a2af471de47f811e1cab88b\nRelated-Bug: #1843043\nStory: 2006952\nTask: 37634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d10fade5a1f38b01f987d5d857a886c173b101d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d10fade5a1f38b01f987d5d857a886c173b101d7"}]},"branch":"refs/heads/master"},"45512cb973d1f03da4ed6285ee4c7f2d390b7970":{"kind":"REWORK","_number":2,"created":"2019-12-04 12:21:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/696129/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/696129/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/696129/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/696129/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/696129/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/696129/2"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 16:24:46.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-04 12:21:19.000000000","tz":0},"subject":"Use mariabackup for database backups","message":"Use mariabackup for database backups\n\nKolla ansible switched its database backup support to mariabackup from\nxtrabackup due to incompatibilities. See\nhttps://bugs.launchpad.net/kolla/+bug/1843043 for details.\n\nChange-Id: Ib95771f09fd6d5e71a2af471de47f811e1cab88b\nRelated-Bug: #1843043\nStory: 2006952\nTask: 37634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45512cb973d1f03da4ed6285ee4c7f2d390b7970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45512cb973d1f03da4ed6285ee4c7f2d390b7970"}]},"branch":"refs/heads/master"},"e93bbd9ed568e98507c792244df58672ab3d84ef":{"kind":"REWORK","_number":3,"created":"2019-12-11 17:17:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/696129/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/696129/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/696129/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/696129/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/696129/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/696129/3"}}},"commit":{"parents":[{"commit":"9332c211f47a71bd214ad94c5d328078d3ccf199","subject":"Merge \"Add kayobe as openstack project for release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9332c211f47a71bd214ad94c5d328078d3ccf199"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-26 16:24:46.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-11 17:16:59.000000000","tz":0},"subject":"Use mariabackup for database backups","message":"Use mariabackup for database backups\n\nKolla ansible switched its database backup support to mariabackup from\nxtrabackup due to incompatibilities. See\nhttps://bugs.launchpad.net/kolla/+bug/1843043 for details.\n\nChange-Id: Ib95771f09fd6d5e71a2af471de47f811e1cab88b\nRelated-Bug: #1843043\nStory: 2006952\nTask: 37634\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e93bbd9ed568e98507c792244df58672ab3d84ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e93bbd9ed568e98507c792244df58672ab3d84ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
