)]}'
{"id":"openstack%2Fkayobe~696371","triplet_id":"openstack%2Fkayobe~master~If432afde374fe247d09c952e110c9567e17daea1","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"If432afde374fe247d09c952e110c9567e17daea1","subject":"Support internal API TLS encryption","status":"MERGED","created":"2019-11-27 17:40:20.000000000","updated":"2019-12-12 13:09:31.000000000","submitted":"2019-12-12 13:07:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696371-1576156062532-19300d8e","meta_rev_id":"7655f31b4749792b751942d186a16303dd5a4ab2","_number":696371,"virtual_id_number":696371,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 13:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-12 11:03:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-12 10:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-12-12 11:03:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-12-12 13:09:31.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 10:59:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-12-12 11:03:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-12 13:07:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c9d48f36c52f7679bd09a442b5de140b5e513608","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-27 17:40:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15dc19a6e636010703e9ebc9b52c3f44fe7c4a17","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-27 17:45:10.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"23baa290b4dfa8fea3934d94255a20efae269854","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-27 17:50:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"50ad10544eda749d31a148e884c5510cb81aa55d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-27 17:56:36.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fb1ef667e4e62b323ad42952f90e81341efbe059","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 19:56:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1156e8d8b32743c69f23d35fa1d84c70 : SUCCESS in 4m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e5a02c004625483d9750542f5613c813 : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2185aa7de954337a89903186da2d82c : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbeb3f1af0624465b0a07023c591031c : SUCCESS in 4m 23s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cf3439f2a0fe43408bb325cc9941b078 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6912a16d2c7941688b76aff123b0f420 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2f10b5480f044b385d7c26eb010e07e : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f1efdc6076c48cd998d3759ead4b632 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/54c31c5199764fca8688d77068347d00 : FAILURE in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b870fb1ecb94b9f86087a55ee0ae876 : SUCCESS in 37m 10s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/361a0123a0f04042a4b4131d299e4f89 : SUCCESS in 1h 12m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0807474fb4444343b1422c1c4c98a10e : SUCCESS in 1h 54m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/a78484461eb74293858f1f5999ac1db5 : SUCCESS in 25m 12s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/770dac3765cb435e8692c821ca30ecd1 : SUCCESS in 43m 23s","accounts_in_message":[],"_revision_number":4},{"id":"6604f9f6088cbd98c9166ce1e2db017658763dfa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-28 15:13:09.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d403464a9a0691cb7769ee43ae91dcc69a0b4285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 17:23:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90582e69a0ac448bb9fe02c18473ef7c : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e340cf9162cd442a8007e076a4f69532 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aacc01ef57f04f54a112cad586bb506e : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8352d49223bb41fa82c05feac01f7736 : SUCCESS in 4m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6b32598e958d4c5eb82cc0fbe2a249ad : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62575667d37b403190409099164acb3f : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03501920b734425788612742a023bc80 : SUCCESS in 3m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ac214c5d2f643c4a401e90e66bab45c : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c60fc96406d34ac7b24b49b0e59545c3 : SUCCESS in 7m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a3f6c86f73b4195b79109b49cbcab0b : SUCCESS in 38m 10s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/23d5ac47630d4263a0341629fd9c33ca : SUCCESS in 58m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c85aa85fc6234f17839550f496d7d3c5 : SUCCESS in 2h 05m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/679b5792e969479eaa74ad56ae096dfc : SUCCESS in 21m 05s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/26594588ee73454eba9addae96a2b004 : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":5},{"id":"a24ac2487148098fbdf8ff8db882cf03d85039ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-06 16:53:24.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"3ccc03e295e83de78ccd852b468baab4fde04350","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-11 16:57:59.000000000","message":"Patch Set 5:\n\nNeeds a rebase","accounts_in_message":[],"_revision_number":5},{"id":"280ee2b383e2a8d17049e1438d31229f1fdb337d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-11 17:13:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0e87217dc5d5d5debcb847cbe63827c10d02256f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 19:18:39.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03ef47146c6a4f15bc9e22afd1465eb6 : SUCCESS in 5m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a4f46121776245da8f8c20e4a428d2c2 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eea58e30e2c6491db8b3c868b01d2cb9 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38614d41769844a1beaffedf28e12eb5 : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8b9ed4732fbb4ef8bcefc7e6ea5b8191 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad748347de924859b73d85bc51869459 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb62eaa69f284f6c9d003dde0236b411 : SUCCESS in 3m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/086bb12dc063499c94e5c1d6e5e7c91c : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6a70b739b1f48c6be2fce66dd086b05 : SUCCESS in 8m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e50f640c6cde47cc964d046bacd0ada5 : SUCCESS in 39m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5161dc978e664051bdb62da2930e5ccf : SUCCESS in 1h 22m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/31a2ff9d9e8d456e90a7a7877fe8483b : SUCCESS in 1h 59m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3ddc9cc0bfa74feba6b602d97f069ffe : SUCCESS in 19m 07s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c0b5195392304495adab65e171e789e1 : SUCCESS in 32m 12s","accounts_in_message":[],"_revision_number":6},{"id":"6c6e5fb2b5fbc1b77b97bb6c50e16b11ebbdc781","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-12 10:59:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7d4ce0ba7deb643aef75d0ece3d8fbc9e5f11606","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-12 11:03:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"cba1ce4837b474071cb5a283cd28ff24c8171679","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 11:03:29.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"48ab70f9e8161b8682886e30c7c895c3b9554b5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:07:42.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6c9a3e445b61485f9e972ecf10f5c729 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bdf09c403b5460e944a39c67accfb98 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086be546ebf743449eb3530a96e83f25 : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1ecc7925781a4997bcc5dc0db33b54b8 : SUCCESS in 8m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8cb08d44b354ff39ac308a8d84ff441 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b414ab959af84f2187b75fbcc819b0cd : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f5e51fcf7ef34bf4a9c30cbf73e74365 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1d8b2bc143c34020b65da369e3f858f4 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f6eddc342044ba888546ad2bd18cf78 : SUCCESS in 39m 08s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/6d04f2e3a0554308badb43c5c8d0ae08 : SUCCESS in 1h 19m 19s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/79aa7b78aced42deaa0819ada4cdfc2f : SUCCESS in 1h 58m 37s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/be5ff05ffa124d4d9364b127ea5f0ed6 : SUCCESS in 19m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/dcea8b774f2f4c14b6176b1a9a9ef9c7 : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":6},{"id":"c8a180e212af0868fa5c705e0eb5b11dd7a178b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:07:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"7e3c72b7dcee6cb724dc99e508aad725c8c9e600","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:09:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da984000284c4a25a639c3fb93a09f9c : SUCCESS in 1m 29s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46ba3aed58864d5899c90647a9c014af : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c8f105a9070e25cee613962683639a351aad185c","revisions":{"cd90e0c64ad301c9799c3a9ea2576a934f0b912c":{"kind":"REWORK","_number":1,"created":"2019-11-27 17:40:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/696371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/696371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/696371/1"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:37:39.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:40:02.000000000","tz":0},"subject":"Support internal API TLS encryption","message":"Support internal API TLS encryption\n\nKolla Ansible Train introduces support for TLS encryption of the\ninternal API. This change introduces support for internal API encryption\nin Kayobe.\n\nThe following new variables are introduced:\n\n* kolla_enable_tls_internal\n* kolla_internal_tls_cert\n* kolla_internal_fqdn_cacert\n\nChange-Id: If432afde374fe247d09c952e110c9567e17daea1\nStory: 2006959\nTask: 37649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd90e0c64ad301c9799c3a9ea2576a934f0b912c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cd90e0c64ad301c9799c3a9ea2576a934f0b912c"}]},"branch":"refs/heads/master"},"4437bc4ce05ef4588549547ce169b24ba1c5f4f8":{"kind":"REWORK","_number":2,"created":"2019-11-27 17:45:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/696371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/696371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/696371/2"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:37:39.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:45:03.000000000","tz":0},"subject":"Support internal API TLS encryption","message":"Support internal API TLS encryption\n\nKolla Ansible Train introduces support for TLS encryption of the\ninternal API. This change introduces support for internal API encryption\nin Kayobe.\n\nThe following new variables are introduced:\n\n* kolla_enable_tls_internal\n* kolla_internal_tls_cert\n* kolla_internal_fqdn_cacert\n\nChange-Id: If432afde374fe247d09c952e110c9567e17daea1\nStory: 2006959\nTask: 37649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4437bc4ce05ef4588549547ce169b24ba1c5f4f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4437bc4ce05ef4588549547ce169b24ba1c5f4f8"}]},"branch":"refs/heads/master"},"5f5eca15dd6ed953950e625aa28f171559340cff":{"kind":"REWORK","_number":3,"created":"2019-11-27 17:50:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/696371/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/696371/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/696371/3"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:37:39.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:49:06.000000000","tz":0},"subject":"Support internal API TLS encryption","message":"Support internal API TLS encryption\n\nKolla Ansible Train introduces support for TLS encryption of the\ninternal API. This change introduces support for internal API encryption\nin Kayobe.\n\nThe following new variables are introduced:\n\n* kolla_enable_tls_internal\n* kolla_internal_tls_cert\n* kolla_internal_fqdn_cacert\n\nAlso only set kolla_*_fqdn_cacert in globals.yml if set.\n\nChange-Id: If432afde374fe247d09c952e110c9567e17daea1\nStory: 2006959\nTask: 37649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f5eca15dd6ed953950e625aa28f171559340cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f5eca15dd6ed953950e625aa28f171559340cff"}]},"branch":"refs/heads/master"},"4e062de2bcb3ab39c5bf2ad7a4c9b6cacc9f1970":{"kind":"REWORK","_number":4,"created":"2019-11-27 17:56:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/696371/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/696371/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/696371/4"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:37:39.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:56:24.000000000","tz":0},"subject":"Support internal API TLS encryption","message":"Support internal API TLS encryption\n\nKolla Ansible Train introduces support for TLS encryption of the\ninternal API. This change introduces support for internal API encryption\nin Kayobe.\n\nThe following new variables are introduced:\n\n* kolla_enable_tls_internal\n* kolla_internal_tls_cert\n* kolla_internal_fqdn_cacert\n\nAlso only set kolla_*_fqdn_cacert in globals.yml if set.\n\nChange-Id: If432afde374fe247d09c952e110c9567e17daea1\nStory: 2006959\nTask: 37649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e062de2bcb3ab39c5bf2ad7a4c9b6cacc9f1970"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4e062de2bcb3ab39c5bf2ad7a4c9b6cacc9f1970"}]},"branch":"refs/heads/master"},"014aead243582969f985ac92fb73ec86dc93e48b":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-11-28 15:13:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/696371/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/696371/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/696371/5"}}},"commit":{"parents":[{"commit":"994d7902439789a63cf42caf1c0102349a2b9fb2","subject":"Fix kolla-ansible role tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/994d7902439789a63cf42caf1c0102349a2b9fb2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:37:39.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-28 10:41:08.000000000","tz":0},"subject":"Support internal API TLS encryption","message":"Support internal API TLS encryption\n\nKolla Ansible Train introduces support for TLS encryption of the\ninternal API. This change introduces support for internal API encryption\nin Kayobe.\n\nThe following new variables are introduced:\n\n* kolla_enable_tls_internal\n* kolla_internal_tls_cert\n* kolla_internal_fqdn_cacert\n\nAlso only set kolla_*_fqdn_cacert in globals.yml if set.\n\nChange-Id: If432afde374fe247d09c952e110c9567e17daea1\nStory: 2006959\nTask: 37649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/014aead243582969f985ac92fb73ec86dc93e48b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/014aead243582969f985ac92fb73ec86dc93e48b"}]},"branch":"refs/heads/master"},"c8f105a9070e25cee613962683639a351aad185c":{"kind":"REWORK","_number":6,"created":"2019-12-11 17:13:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/71/696371/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/696371/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/696371/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/696371/6"}}},"commit":{"parents":[{"commit":"9332c211f47a71bd214ad94c5d328078d3ccf199","subject":"Merge \"Add kayobe as openstack project for release notes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9332c211f47a71bd214ad94c5d328078d3ccf199"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-27 17:37:39.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-12-11 17:13:38.000000000","tz":0},"subject":"Support internal API TLS encryption","message":"Support internal API TLS encryption\n\nKolla Ansible Train introduces support for TLS encryption of the\ninternal API. This change introduces support for internal API encryption\nin Kayobe.\n\nThe following new variables are introduced:\n\n* kolla_enable_tls_internal\n* kolla_internal_tls_cert\n* kolla_internal_fqdn_cacert\n\nAlso only set kolla_*_fqdn_cacert in globals.yml if set.\n\nChange-Id: If432afde374fe247d09c952e110c9567e17daea1\nStory: 2006959\nTask: 37649\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8f105a9070e25cee613962683639a351aad185c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c8f105a9070e25cee613962683639a351aad185c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
