)]}'
{"id":"openstack%2Fkayobe~696488","triplet_id":"openstack%2Fkayobe~master~I43d4761409df3eea146dc7ff16b4c8d80855a91b","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I43d4761409df3eea146dc7ff16b4c8d80855a91b","subject":"Fix kolla-ansible role tests","status":"MERGED","created":"2019-11-28 10:41:55.000000000","updated":"2019-12-06 15:03:12.000000000","submitted":"2019-11-28 15:50:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696488-1574956220140-d86a0cf6","meta_rev_id":"c654029d462bdd1fef702ee654639da812fd33f4","_number":696488,"virtual_id_number":696488,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-28 15:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-11-28 13:37:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-11-28 13:49:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2019-11-28 13:49:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-11-28 13:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-11-28 13:32:08.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-28 13:32:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-11-28 13:37:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-11-28 13:49:50.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-11-28 15:50:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"aaed1446d6de41298411b59f1496689f62ec560e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-28 10:41:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0a6f836f54472f2658f316b2ad9306039b64b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 12:44:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7aad20e579134564ad70af314a2bc641 : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/096f113fde854dd3b9aac409ad04c4ca : SUCCESS in 3m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2de47f5e458e4303b0eddfb90daefae3 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f028290ff69493eb2338f70b47a9f84 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fba4bf8b6f14b3fb241accff93605c3 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e771d049b85e4a8c8bfad9c5a4731681 : SUCCESS in 6m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d2abb96a65dc46b599d0ab2326e7ed91 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17a4656619414699ab6fde70c5b46178 : SUCCESS in 9m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/47175b6be1ad44a5baf48b57fb5a3264 : SUCCESS in 29m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9806e1e63a0744928b0a9ebbe0006a6a : SUCCESS in 1h 17m 49s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/91b89f4f3bc54c4db57ac41589fe85d9 : SUCCESS in 2h 01m 10s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/a87c650d960b42efbc5216c5a7e5b8c6 : SUCCESS in 23m 47s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b2dc80f4c5b04ff3b33e5504442e0482 : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"cefdf254bfdd7f787a2552bc6462f5cd4f637a1c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-11-28 13:31:46.000000000","message":"Patch Set 1: Review-Priority+2\n\nBlocking CI","accounts_in_message":[],"_revision_number":1},{"id":"3a01e9a08bf326342b8d1572adbaaca67120717c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-11-28 13:37:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1ab6c00acc4b9b82f505b6b611bcc9648ebd794e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-11-28 13:49:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7bef446ca03c186f788aedd37127084479a8c56","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 13:50:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b8640ff55610901a044cf2f113ac71fb29a3b7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 15:50:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1faaccaf5dc34e259166f35e0339b893 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ef58bcc27154b6cb060e939b5cb14e4 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f236a2b8a71744b8888568b506e34691 : SUCCESS in 5m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4f869644eff84188bc3314b201169136 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f725dbb903d411885115cf6ba041d0d : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b9b31d8ec4b4abda7eef85257708f0d : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e931f7bb7afb4deea18c10ea6e079a6c : SUCCESS in 8m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6f64af49a5ac4208bfc6188f30df1fa8 : SUCCESS in 38m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/aaccab14d22a415abcd8fa6a6c261cd2 : SUCCESS in 1h 11m 47s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c30082b103064f3c92804e3317db8a66 : SUCCESS in 1h 54m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c91aef78a88c420e9fce9c6512e8203f : SUCCESS in 21m 30s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d9608d0f9be148638c48c2b0fc80816e : SUCCESS in 43m 15s","accounts_in_message":[],"_revision_number":1},{"id":"bb29e96d454c4761ba41d3d3091d86e40d20d29a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 15:50:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"30c708a59ac5a9fe64b4cb05e0b31b360c8688a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 15:51:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f06e91ac0b8f4a339103b6f487131ee9 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"c654029d462bdd1fef702ee654639da812fd33f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-06 15:03:12.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit a19566dc40dd9f1e0c3ff474bf527fd46d9368eb","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"994d7902439789a63cf42caf1c0102349a2b9fb2","revisions":{"994d7902439789a63cf42caf1c0102349a2b9fb2":{"kind":"REWORK","_number":1,"created":"2019-11-28 10:41:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/88/696488/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/88/696488/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/696488/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/696488/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/88/696488/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/88/696488/1"}}},"commit":{"parents":[{"commit":"a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6","subject":"Merge \"Document requirements for BIOS and RAID configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a30cf063d15323c0c30ecf3aa9a4d3b06d21aac6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-28 10:39:41.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-11-28 10:41:08.000000000","tz":0},"subject":"Fix kolla-ansible role tests","message":"Fix kolla-ansible role tests\n\nThe kolla-ansible CLI now requires the kolla_ansible python module to be\nimportable. We need to activate the virtualenv to make this work.\n\nChange-Id: I43d4761409df3eea146dc7ff16b4c8d80855a91b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/994d7902439789a63cf42caf1c0102349a2b9fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/994d7902439789a63cf42caf1c0102349a2b9fb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
