)]}'
{"id":"openstack%2Fkayobe~697684","triplet_id":"openstack%2Fkayobe~master~I074d4b9d444649ecf956d3cd92748862e8c89a5c","project":"openstack/kayobe","branch":"master","topic":"tox-ansible-branch","hashtags":[],"change_id":"I074d4b9d444649ecf956d3cd92748862e8c89a5c","subject":"Use {{ openstack_branch }} as version of kolla-ansible in ansible tests","status":"MERGED","created":"2019-12-06 16:06:49.000000000","updated":"2019-12-12 12:41:16.000000000","submitted":"2019-12-12 12:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697684-1576154341123-8734d19d","meta_rev_id":"9cb3157531842781af17a48cdd91f32ffba175f0","_number":697684,"virtual_id_number":697684,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 12:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2019-12-12 10:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2019-12-09 10:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2019-12-12 10:44:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-09 10:03:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-12 10:44:07.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2019-12-12 12:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7b30850e03c50a1e77d1460e1a45848cfa0f1279","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-06 16:06:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b25cca6af569466d956ff5b840376c3e7889bf7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-06 16:30:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c390f5ca63e1ecc985838c081b679b2722a6bcaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:29:28.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ca7b46c9877438593a7036a79cf0a3e : SUCCESS in 5m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a7f339a23a044f32b2e6595a650266ab : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f60026831fce418baddd63e3061bfcb4 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75974bcdbd754154838dfd197f50bbcc : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d65cf81f13ec4577a202deabdd4c0f38 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ddd7ebca9d54a5eb35eceb39217e5ad : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd222adec60b4200b2e7bd2c102094fa : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/892c3cb67fe14fafadddfed53de37c65 : SUCCESS in 9m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1319ae5395764d96a58f7596b8534596 : SUCCESS in 36m 37s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/168c343b3b6546ce84aeff8f617c12d8 : SUCCESS in 1h 12m 13s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a49f3ddb8df54ea4a0719354da5f8f84 : SUCCESS in 1h 53m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3f413316c7be48899131fb21489fa9ad : SUCCESS in 25m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e4adfe8ae47045fead294ad7c1e059f7 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":2},{"id":"618252a958bd1da3c1dc2375229b45d192ff7651","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-09 10:03:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"148fd5bef3df9b0d2d77742ed5cb88dfc6057580","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-12-12 10:44:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6cff95379b5be0524890d730d52ac5c6bf72ccfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 10:44:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"11bb507765cd4859e8bddf736e9adc3f3001f072","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:39:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7d1a55294eb84377b52680dc1896d405 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/145267891c2441d1bfb6cf283f0f4bd2 : SUCCESS in 7m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3fd1ebf935b44bd8698b936faa38550 : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/616996212660482389ff13abfd3d9635 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2858dcff59b64116b924f6fd0f5be948 : SUCCESS in 8m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b18442f2e336454d9fbb1fd0cd75cc89 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b2e947f7d024f0eb40c58c89c13c4d3 : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b4c5fe217bd44d792435d0d54c7e558 : SUCCESS in 39m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/24097a491f6f4c9c9350c95409e1e21a : SUCCESS in 1h 15m 02s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/86556f909ce6471a9a8408e1eaf29e68 : SUCCESS in 1h 50m 00s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6dcb175f328c4ae9a2eaaca19b02e78f : SUCCESS in 22m 59s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/cd9639ebfe57420b85e3613d615b2557 : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":2},{"id":"ae9244760c945d6376227786d19861c700b45c9f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:39:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"62705f7bb57c2d0941b5044ebfea0f71316d78b2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 12:41:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76922e86e4a740209fa7058f725a39de : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"225130a594e84776b3bfcc852da37cbaeb867c42","revisions":{"e4bc60f6fd711d63a36c2299c56fa58bd71e6ed7":{"kind":"REWORK","_number":1,"created":"2019-12-06 16:06:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/697684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/697684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/697684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/697684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/697684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/697684/1"}}},"commit":{"parents":[{"commit":"3bae0fd2e771f530e43b253c68711f6ef789dd84","subject":"Merge \"Remove workaround for kolla-ansible post-deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bae0fd2e771f530e43b253c68711f6ef789dd84"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-06 16:05:46.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-06 16:05:46.000000000","tz":60},"subject":"Use {{ openstack_branch }} as version of kolla-ansible in ansible tests","message":"Use {{ openstack_branch }} as version of kolla-ansible in ansible tests\n\nInstead of always checking out the master branch, use the kolla-ansible\nbranch with which this Kayobe version is meant to be used.\n\nChange-Id: I074d4b9d444649ecf956d3cd92748862e8c89a5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4bc60f6fd711d63a36c2299c56fa58bd71e6ed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e4bc60f6fd711d63a36c2299c56fa58bd71e6ed7"}]},"branch":"refs/heads/master"},"225130a594e84776b3bfcc852da37cbaeb867c42":{"kind":"REWORK","_number":2,"created":"2019-12-06 16:30:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/84/697684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/697684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/697684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/697684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/697684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/697684/2"}}},"commit":{"parents":[{"commit":"3bae0fd2e771f530e43b253c68711f6ef789dd84","subject":"Merge \"Remove workaround for kolla-ansible post-deploy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bae0fd2e771f530e43b253c68711f6ef789dd84"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-06 16:29:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-06 16:29:45.000000000","tz":60},"subject":"Use {{ openstack_branch }} as version of kolla-ansible in ansible tests","message":"Use {{ openstack_branch }} as version of kolla-ansible in ansible tests\n\nInstead of always checking out the master branch, use the kolla-ansible\nbranch with which this Kayobe version is meant to be used.\n\nChange-Id: I074d4b9d444649ecf956d3cd92748862e8c89a5c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/225130a594e84776b3bfcc852da37cbaeb867c42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/225130a594e84776b3bfcc852da37cbaeb867c42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
