)]}'
{"id":"openstack%2Fkayobe~698882","triplet_id":"openstack%2Fkayobe~master~I2695034d36936fcc77a4828c67f9552155781dd6","project":"openstack/kayobe","branch":"master","topic":"story/2004291","hashtags":[],"change_id":"I2695034d36936fcc77a4828c67f9552155781dd6","subject":"Support custom Kolla group_vars","status":"MERGED","created":"2019-12-13 11:28:32.000000000","updated":"2019-12-20 11:39:10.000000000","submitted":"2019-12-20 11:37:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698882-1576841849672-17819b5d","meta_rev_id":"2bf2eae857d1255c8f215da081af476bae30106c","_number":698882,"virtual_id_number":698882,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-18 17:31:19.000000000","_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 11:37:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":2,"date":"2019-12-19 12:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-18 17:12:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":1,"date":"2019-12-19 12:22:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-20 08:07:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 21:36:21.000000000","updated_by":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"reviewer":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"state":"REVIEWER"},{"updated":"2019-12-19 12:22:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-20 08:07:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-12-20 11:37:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"37dad6dd01cada9904c5f93b24ec9be6998feb0e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-13 11:28:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5f816441127d48f2766b82a499a28b94bb2d64d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-13 12:17:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"410cfa32dc597011cc7a6e7978b0271b48018ff9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 13:14:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5b9b9e3c0a2a4e1dbb0e09c3964b204c : SUCCESS in 4m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0a7dc99858f243f48fd27b0797b1eb7c : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54b3a4b3a9654a6c84da79d8abc80be4 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec8ca79018b4f60819834e94bc76a9d : FAILURE in 4m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f513e56fdf614f46b99019c716b5c8fb : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28bcf8976dbd42979f2c9c8c74162035 : FAILURE in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f71a9776d294d22948977c62a10c514 : SUCCESS in 5m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cff66b2e064a4ddfa81cc3f3fb1c12da : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/648dacb4f09a4332a6da4e4c325a59d1 : FAILURE in 7m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/05bdf9348ec2426b92748d753c2aa48b : SUCCESS in 31m 10s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8df7b2344ffe47be9769507a6034034e : SUCCESS in 1h 13m 20s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3f929d6a1a3f4eddb6e981838549c7d2 : SUCCESS in 1h 37m 37s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5680f7970b634d62bfb253cfee91df7c : SUCCESS in 22m 49s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e5984a1591db445b88395f6e3369d17a : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":1},{"id":"641ea87991d092a44c84766380976e165fb00607","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 14:09:59.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cc9de9ddc03941a7aa87b7d61954938f2465e7f6","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-13 14:12:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd1c76b78ffbfacbe00cec233d9ab4c5e33ca77d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-13 14:12:26.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"fa9ea7d7906f7a55a6d146f179403c81661522ac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 14:19:02.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e1a05d8856cbab2010422c3992b4418183e50b28","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-13 15:13:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e5220de8afc1bba5d9a9a4f18883f860542cb9c9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-13 15:15:05.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nCould use a test as well, will circle back to that after pushing related patch.","accounts_in_message":[],"_revision_number":1},{"id":"a0937c28b0541ef28e4304460f4502f02b192bdd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 16:03:12.000000000","message":"Patch Set 3: Code-Review+1\n\nLooks good overall. A test would be nice.","accounts_in_message":[],"_revision_number":3},{"id":"322557cbf3af865cf5f7c67176506621b6bea913","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-13 16:28:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"85e637d86d0f5596b41be5ce3161da4a05da5b86","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-13 16:28:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"abd3a7c5df67ed11800e0ce23568587fbcbd006e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 16:38:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c48723355407690e012a9ffc753f2855be52f2c5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-13 17:03:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eea0454a42f4c10082768718813f11698c30a90b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:10:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45d1277697f942daa5974666f6585367 : SUCCESS in 6m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a5bdc5d785634e94b06f01f6f392b325 : SUCCESS in 8m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09f380ae67b14749b9e81fa61b2791b6 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94a5fbbd1c214f40b56c9a933f4cead8 : FAILURE in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/241164f4cc8c4ea786c935916c61d17b : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51163b653d914137971ac226458ad238 : FAILURE in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/111e92ddaf3847489e0386422968cc73 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/96f2d89b15844499938d87ef34f42bd6 : SUCCESS in 7m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09df666a12cc4793a7f186e46b2ac8d3 : FAILURE in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94336e164c634bd78f045601b3f66bc5 : SUCCESS in 38m 46s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/21b377d53eec490f946f8fa6d5d1a377 : SUCCESS in 1h 19m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/79dcae52e574436e8d3e23456b261a26 : SUCCESS in 1h 51m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e52cc2d032f846b2aaddd22ab70f6248 : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/163bbe3c97ec4f4b8e0e7730b7793732 : SUCCESS in 34m 09s","accounts_in_message":[],"_revision_number":3},{"id":"566c4d39bd4f37445586a02b857543a0995c6284","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-13 17:10:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5607408f25f031afd6ec247bf70ec1b18f91f3ed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-13 17:12:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9e49fe953e45f11ff308536f9051f10bdfffbf2c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-13 17:34:56.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"5aea4e37151e5a23e0afbb91db27b02738e94425","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 15:53:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"77125a1add1e4402ca2c260407c783a7e9c3b096","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 15:57:35.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"33ba40d30fb66207b2554c72d9a1f6232fd81cb2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 16:13:56.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nThanks, LGTM.","accounts_in_message":[],"_revision_number":5},{"id":"148963b76ee72f8fbbe51952daa90e3ce39158b0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 17:01:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"bc6f94d83c81254dcfe2c9f86c609b0eb9dc66cf","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 17:02:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7bda040a1c087b2eaea24d81053752c36bd3f5b3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 17:10:27.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"ba9b1303f4463f083d45824bd15232bd5cbe492c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 18:50:40.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5adf493968f4f6eaf15f7e01a7f093e : SUCCESS in 5m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/927070827df04996a41301d272490f04 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c83f4c035094b04968fc90987467a7c : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170804faba4e48a9a7746dd67ada63e1 : SUCCESS in 5m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45c633c879494637abf2352e6f9c5ead : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caf38900d4a54645bc339d0152b93b8c : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5652e5aaba30470b9cdd936ed1de85c5 : SUCCESS in 5m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/db2f545a7bf24b6fb98fdc505d790a7e : SUCCESS in 6m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e921cfd44cd542cab2f9342c7c5cb0a1 : SUCCESS in 9m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2314069004704adc925295148b8d0dc1 : SUCCESS in 42m 02s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/755ad056416b4f3f81a14aedf883c764 : SUCCESS in 1h 20m 25s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a713c76b83cd4731882adc1056f7e0bf : SUCCESS in 1h 47m 54s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0959aa1750154dc0aa90af77f810aa98 : SUCCESS in 22m 29s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4f40997f883e424d91abb5926f3fed9a : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":6},{"id":"5a190864bd26f2b359633da3b1e5d3df5b7645b9","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2019-12-17 21:36:21.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"4ca2b75a477b1fc81fc871b7b876bfc908ba6af3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 15:46:32.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3ae58b00b6c57ff8b08cbd14c1766b0fee59f219","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 17:12:26.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a9bd6d740d11f943fd9dafa59d749732b32836e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 17:25:52.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c11b6e97d9774c8ca93c722c89fa4ccc : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1fcef2906167416b82591713c023584a : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/803de269b2f641108019a102ea54d905 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72417def329c4aaeb5b2fdd6bdcff553 : SUCCESS in 6m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ee854a47112a4f6e849b6b4c39ef57b7 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9bb993206a424d98120b9ffa3763ce : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26e70f693d1d4029af74aeda47973674 : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f44ba77c6ecf44a5b3fbbea11686b8b5 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6697c67e8bf74f9690eda86242a47e7f : SUCCESS in 9m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9f730a8539a44aa193395ea8f908fce4 : SUCCESS in 33m 13s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/58b20c394e6742dab43bcd1cf8f2b7ce : SUCCESS in 1h 11m 43s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/382a92373aa64a03b8a28d7b226a4012 : SUCCESS in 1h 35m 56s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ecddedd304e0434ab9272737237d28b5 : FAILURE in 17m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0f8bc022a91841d0bdef4a443105a8fe : FAILURE in 39m 20s","accounts_in_message":[],"_revision_number":7},{"id":"576d95249a149043d915f591b1d15462f2675f55","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-18 17:31:19.000000000","message":"Patch Set 7:\n\nrecheck Failed adding an inspector rule - service failed to come up?","accounts_in_message":[],"_revision_number":7},{"id":"78f8040c1f60e1c4c3b15057dabd349191e6e4fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 19:37:03.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12c6fa237c3b4e0f95764b558c7b6c89 : SUCCESS in 6m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/61450c69b88647ce9f424c25e5fabf19 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b8eb4d2b298461c90f5cc1a6ba5ca42 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/301f631a0a1a4ee196a42f70bbb7155b : SUCCESS in 4m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1b990f42bb2441697cdf73f056dcc2c : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a5bbc70834249ec954addf8e5fbe9cd : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faeee99dab5041bb8223693cbbaca3d1 : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5a9687539324a53bcc55d5d2c270302 : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ea16b362294640729d2e8158f037bc5a : SUCCESS in 9m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c6c5828b6ec1487f8d04b6aecd412abf : SUCCESS in 37m 33s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/354e87dab3bf4ae8ab4e0cfd28ac254d : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b64b9bc6dd804964ba12be62489c39c9 : SUCCESS in 1h 59m 51s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8488533df1fe4bd580fe99dc2b1b59ae : FAILURE in 23m 59s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0c8dd9d836544262abf1e648b553b31f : FAILURE in 35m 01s","accounts_in_message":[],"_revision_number":7},{"id":"03c0a23b43f969b7facfe9d7e25d87d1dce46656","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 21:09:39.000000000","message":"Patch Set 7:\n\nThis is caused by the release of ironic-lib 3.0.0. See https://dafcf7323a7c828b67e0-814491f7a8e968b50eeaf14b0c7f115f.ssl.cf1.rackcdn.com/698882/7/check/kayobe-seed-centos/8488533/primary/kolla/ironic-api/ironic-api-journal.txt.gz","accounts_in_message":[],"_revision_number":7},{"id":"4f9cbff8b174ea40df7a57320465b0d4e148a159","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 21:48:09.000000000","message":"Patch Set 7:\n\nI think the problem is that the image is using bifrost-7.0.0, which is fetching from master branches, including for the requirements repo… which has bumped its ironic-lib upper-constraints earlier today: https://review.opendev.org/#/c/698981/\n\nThis was fixed later in stable/train but a new version was never released. Though I don\u0027t understand why 7.0.0 ends up used, the source container image should be using the latest stable/train.","accounts_in_message":[],"_revision_number":7},{"id":"1294d7e51d7847524b0d9c95c060b19d80aa1e41","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 21:52:06.000000000","message":"Patch Set 7:\n\nConfirmed:\n\ndocker pull kolla/centos-source-bifrost-deploy:train\ntrain: Pulling from kolla/centos-source-bifrost-deploy\nDigest: sha256:1d0727af563cb15f9777d8528cba5d4fc70d236a722136331e5965550d580f35\nStatus: Image is up to date for kolla/centos-source-bifrost-deploy:train\ndocker.io/kolla/centos-source-bifrost-deploy:train\n\ndocker run --rm -it kolla/centos-source-bifrost-deploy:train grep ^git_branch playbooks/roles/bifrost-prep-for-install/defaults/main.yml\ngit_branch: master","accounts_in_message":[],"_revision_number":7},{"id":"a68db36468223777039aa0e6c9a3e4cb904dbe12","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 22:15:57.000000000","message":"Patch Set 7:\n\n\u003e I think the problem is that the image is using bifrost-7.0.0, which\n \u003e is fetching from master branches, including for the requirements\n \u003e repo… which has bumped its ironic-lib upper-constraints earlier\n \u003e today: https://review.opendev.org/#/c/698981/\n \u003e \n \u003e This was fixed later in stable/train but a new version was never\n \u003e released. Though I don\u0027t understand why 7.0.0 ends up used, the\n \u003e source container image should be using the latest stable/train.\n\nReplying to myself, source container use tarballs, see kolla/common/config.py:\n\n    \u0027bifrost-base\u0027: {\n        \u0027type\u0027: \u0027url\u0027,\n        \u0027location\u0027: (\u0027$tarballs_base/bifrost/\u0027\n                     \u0027bifrost-7.0.0.tar.gz\u0027)},","accounts_in_message":[],"_revision_number":7},{"id":"2d622295f7a21c31f9e353dd17b8304648805385","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-19 12:22:48.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"9aebdd1ba8b056fc74e584201ac8a8c2a82af256","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-19 14:12:58.000000000","message":"Patch Set 7:\n\nWe need to wait until the new Bifrost image is on Docker Hub.","accounts_in_message":[],"_revision_number":7},{"id":"d7661a69f1068d41d6e4c2280d0c49627251685c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 14:33:49.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba89cf4f950b4915a948069fe860c6a4 : SUCCESS in 6m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d704edf6bfd94251b697897a28c91129 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d67b7e2e1e6b4db399ae3c47e63b8daa : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c8b51edc994227ac3a9480c3b48197 : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2bf1721fb80946c29d4b6c6588d8e89a : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4c5d92fc9f84a9bbf859d5f08aaa8c5 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1425c85c74d648eb868f54f73f79abfe : SUCCESS in 4m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/57a85efbadaa4caea05cd8351b6bf7d7 : SUCCESS in 5m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2097a0512a5a49eb804ee4736cbc2e05 : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c79c92e88a74cd78b563f71700dc6c6 : SUCCESS in 41m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ff538d4c60274e4fb7bbeea867064650 : SUCCESS in 1h 27m 49s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/bce898466aa54a9d82532e7913917ea7 : SUCCESS in 2h 07m 05s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/844d8ab0116543eeb208cdd15ad3f11f : FAILURE in 20m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/656fb62d1a2b49d8a301af0d1c600bfd : FAILURE in 30m 49s","accounts_in_message":[],"_revision_number":7},{"id":"ac08f498b0e2aa9973fca54767180e2d14b326c0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-20 08:07:43.000000000","message":"Patch Set 7: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"ef066ca5224ff181067de0c2660138a91af39056","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 09:53:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49491b05860042ccb1dcb4fd1c9544f5 : SUCCESS in 4m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0b1d490a376f44499369110977f33a2d : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2be751577f0448dfbef50f1add08353f : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d62ec0482c74878a3a5487b3be6c1b4 : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b77c586a4f9c41f89759f479aa770d49 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6a40d530e0492a9ae3fcc97b4d222d : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37f31fadc4b44e0f8cd0532acf9a7123 : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ae4e4970d0e48e99014de9ee2101dc5 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1b6f7b7b9234401a2100289fb4fe459 : SUCCESS in 8m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aa57fa39b53c4793b6509bbb35ba8ed5 : SUCCESS in 31m 08s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5085e6164665469f8752735db1a29440 : SUCCESS in 1h 06m 40s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b5a9d13bfdd44f639b8c2066488d0f94 : SUCCESS in 1h 42m 50s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/2dff6014273148fdaf1692f860a59760 : SUCCESS in 20m 31s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4ad9db374e514a1d81d766df9e9e1657 : SUCCESS in 35m 22s","accounts_in_message":[],"_revision_number":7},{"id":"32615a45cbcb29ad88b105133c2ddba81dfcf3f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 09:53:29.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"8420a569f1b36caeddabbbc3e11661ce6db578b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 11:37:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3536bc599f1e42bfa489c4e1efbf566c : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c0fa740dc564b4b9c5348eef8ed9a6a : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbcd1bc0ce1d47a2b673de91980375ac : SUCCESS in 5m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6490fbf8955441e6b0ecb6415e68ef65 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67fe7c5a48a34297bc1490233fc853c6 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af9bb26b352244b5a4cf4f67a414c31a : SUCCESS in 3m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7cabb9aed7c3438c9953b69e713722fd : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/293609e8663f47c7849da0602df3c1ca : SUCCESS in 8m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4d4afd226f804e5d970893a2b0cdba90 : SUCCESS in 31m 11s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/1db76f88563342a5abd64138c2778802 : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c3986e9bd52b498ba73d9d8731008587 : SUCCESS in 1h 43m 07s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6cb3c9cbfeb842a6bb9e4d0dc22c2869 : SUCCESS in 20m 29s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5f4ac1d8e80b450c8582910c6765b53e : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":7},{"id":"926637ab101338e2bab947567b1c6f8eb6999349","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 11:37:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"14aaedcabc12d2cee27a6eb9d322c5e0c547626c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 11:39:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6141ad62d30742ffb947a76145663cbe : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71a843a611454913bdd18c4a2469a5d2 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"72d4d64609030201ccdc753b98091552638b03e5","revisions":{"848ecd99a21c62dd9b92e4e5c1051ff54a85e8cb":{"kind":"REWORK","_number":1,"created":"2019-12-13 11:28:32.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/82/698882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/698882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/698882/1"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 09:49:33.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 11:27:46.000000000","tz":0},"subject":"Support custom Kolla group_vars","message":"Support custom Kolla group_vars\n\nIn Kayobe hosts which are part of a Nova cell can be managed via the existing\ncontroller and compute groups. However, since Nova Cells are configured via group\nvars in Kolla Ansible we need some way of setting these. We could pass vars\nthrough to Kolla Ansible host vars using\n`kolla_overcloud_inventory_pass_through_host_vars` but the list of variables\nwhich may be set on a per cell basis is large and undefined.\n\nThis change allows the user to directly specify Kolla Ansible group vars as\npart of Kayobe config, allowing the deployment of Nova Cells by Kayobe to be\nlargely unchanged from the procedure documented in Kolla Ansible.\n\nChange-Id: I2695034d36936fcc77a4828c67f9552155781dd6\nStory: 2004291\nTask: 37804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/848ecd99a21c62dd9b92e4e5c1051ff54a85e8cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/848ecd99a21c62dd9b92e4e5c1051ff54a85e8cb"}]},"branch":"refs/heads/master"},"d94c3b6dcec163be876888e27290bc2135b6c9c2":{"kind":"REWORK","_number":2,"created":"2019-12-13 14:12:01.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/82/698882/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/698882/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/698882/2"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 09:49:33.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 13:40:50.000000000","tz":0},"subject":"Support custom Kolla group_vars","message":"Support custom Kolla group_vars\n\nIn Kayobe hosts which are part of a Nova cell can be managed via the existing\ncontroller and compute groups. However, since Nova Cells are configured via group\nvars in Kolla Ansible we need some way of setting these. We could pass vars\nthrough to Kolla Ansible host vars using\n`kolla_overcloud_inventory_pass_through_host_vars` but the list of variables\nwhich may be set on a per cell basis is large and undefined.\n\nThis change allows the user to directly specify Kolla Ansible group vars as\npart of Kayobe config, allowing the deployment of Nova Cells by Kayobe to be\nlargely unchanged from the procedure documented in Kolla Ansible.\n\nChange-Id: I2695034d36936fcc77a4828c67f9552155781dd6\nStory: 2004291\nTask: 37804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d94c3b6dcec163be876888e27290bc2135b6c9c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d94c3b6dcec163be876888e27290bc2135b6c9c2"}]},"branch":"refs/heads/master"},"56b355c337be7f1913283f1e62a7e90fbc95d047":{"kind":"REWORK","_number":3,"created":"2019-12-13 15:13:30.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/82/698882/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/698882/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/698882/3"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 09:49:33.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 15:11:43.000000000","tz":0},"subject":"Support custom Kolla group_vars","message":"Support custom Kolla group_vars\n\nIn Kayobe hosts which are part of a Nova cell can be managed via the existing\ncontroller and compute groups. However, since Nova Cells are configured via group\nvars in Kolla Ansible we need some way of setting these. We could pass vars\nthrough to Kolla Ansible host vars using\n`kolla_overcloud_inventory_pass_through_host_vars` but the list of variables\nwhich may be set on a per cell basis is large and undefined.\n\nThis change allows the user to directly specify Kolla Ansible group vars as\npart of Kayobe config, allowing the deployment of Nova Cells by Kayobe to be\nlargely unchanged from the procedure documented in Kolla Ansible.\n\nChange-Id: I2695034d36936fcc77a4828c67f9552155781dd6\nStory: 2004291\nTask: 37804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56b355c337be7f1913283f1e62a7e90fbc95d047"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56b355c337be7f1913283f1e62a7e90fbc95d047"}]},"branch":"refs/heads/master"},"164c93e8988ce5d8c8fb0e07aac5e063f9b81e2d":{"kind":"REWORK","_number":4,"created":"2019-12-17 15:53:17.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/82/698882/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/698882/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/698882/4"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 09:49:33.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 15:52:45.000000000","tz":0},"subject":"Support custom Kolla group_vars","message":"Support custom Kolla group_vars\n\nIn Kayobe hosts which are part of a Nova cell can be managed via the existing\ncontroller and compute groups. However, since Nova Cells are configured via group\nvars in Kolla Ansible we need some way of setting these. We could pass vars\nthrough to Kolla Ansible host vars using\n`kolla_overcloud_inventory_pass_through_host_vars` but the list of variables\nwhich may be set on a per cell basis is large and undefined.\n\nThis change allows the user to directly specify Kolla Ansible group vars as\npart of Kayobe config, allowing the deployment of Nova Cells by Kayobe to be\nlargely unchanged from the procedure documented in Kolla Ansible.\n\nChange-Id: I2695034d36936fcc77a4828c67f9552155781dd6\nStory: 2004291\nTask: 37804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/164c93e8988ce5d8c8fb0e07aac5e063f9b81e2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/164c93e8988ce5d8c8fb0e07aac5e063f9b81e2d"}]},"branch":"refs/heads/master"},"1b85c53708605fea9c3e3a6562930ecb02011364":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2019-12-17 15:57:35.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/82/698882/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/698882/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/698882/5"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 09:49:33.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 15:56:28.000000000","tz":0},"subject":"Support custom Kolla group_vars","message":"Support custom Kolla group_vars\n\nIn Kayobe hosts which are part of a Nova cell can be managed via the\nexisting controller and compute groups. However, since Nova Cells are\nconfigured via group vars in Kolla Ansible we need some way of\nsetting these. We could pass vars through to Kolla Ansible host vars\nusing `kolla_overcloud_inventory_pass_through_host_vars` but the list\nof variables which may be set on a per cell basis is large and\nundefined.\n\nThis change allows the user to directly specify Kolla Ansible group\nvars as part of Kayobe config, allowing the deployment of Nova Cells\nby Kayobe to be largely unchanged from the procedure documented in\nKolla Ansible.\n\nChange-Id: I2695034d36936fcc77a4828c67f9552155781dd6\nStory: 2004291\nTask: 37804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b85c53708605fea9c3e3a6562930ecb02011364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b85c53708605fea9c3e3a6562930ecb02011364"}]},"branch":"refs/heads/master"},"ac7e17f3b22b81f3f943c48b5b5e9a3e394b083b":{"kind":"REWORK","_number":6,"created":"2019-12-17 17:01:51.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/82/698882/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/698882/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/698882/6"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 09:49:33.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 17:01:16.000000000","tz":0},"subject":"Support custom Kolla group_vars","message":"Support custom Kolla group_vars\n\nIn Kayobe hosts which are part of a Nova cell can be managed via the\nexisting controller and compute groups. However, since Nova Cells are\nconfigured via group vars in Kolla Ansible we need some way of\nsetting these. We could pass vars through to Kolla Ansible host vars\nusing `kolla_overcloud_inventory_pass_through_host_vars` but the list\nof variables which may be set on a per cell basis is large and\nundefined.\n\nThis change allows the user to directly specify Kolla Ansible group\nvars as part of Kayobe config, allowing the deployment of Nova Cells\nby Kayobe to be largely unchanged from the procedure documented in\nKolla Ansible.\n\nChange-Id: I2695034d36936fcc77a4828c67f9552155781dd6\nStory: 2004291\nTask: 37804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac7e17f3b22b81f3f943c48b5b5e9a3e394b083b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac7e17f3b22b81f3f943c48b5b5e9a3e394b083b"}]},"branch":"refs/heads/master"},"72d4d64609030201ccdc753b98091552638b03e5":{"kind":"REWORK","_number":7,"created":"2019-12-18 15:46:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/82/698882/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/698882/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/698882/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/698882/7"}}},"commit":{"parents":[{"commit":"f6a93bbc2120424a2626b1db96030e42bfdcff01","subject":"Merge \"Use OpenStack Train release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a93bbc2120424a2626b1db96030e42bfdcff01"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 09:49:33.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-18 15:45:36.000000000","tz":60},"subject":"Support custom Kolla group_vars","message":"Support custom Kolla group_vars\n\nIn Kayobe hosts which are part of a Nova cell can be managed via the\nexisting controller and compute groups. However, since Nova Cells are\nconfigured via group vars in Kolla Ansible we need some way of setting\nthese. We could pass vars through to Kolla Ansible host vars using\n`kolla_overcloud_inventory_pass_through_host_vars` but the list of\nvariables which may be set on a per cell basis is large and undefined.\n\nThis change allows the user to directly specify Kolla Ansible group vars\nas part of Kayobe config, allowing the deployment of Nova Cells by\nKayobe to be largely unchanged from the procedure documented in Kolla\nAnsible.\n\nChange-Id: I2695034d36936fcc77a4828c67f9552155781dd6\nStory: 2004291\nTask: 37804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72d4d64609030201ccdc753b98091552638b03e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72d4d64609030201ccdc753b98091552638b03e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
