)]}'
{"id":"openstack%2Fkayobe~698970","triplet_id":"openstack%2Fkayobe~master~Ife90d839a5184a3b692c49e0411cd1ac6467f25a","project":"openstack/kayobe","branch":"master","topic":"story/2004291","hashtags":[],"change_id":"Ife90d839a5184a3b692c49e0411cd1ac6467f25a","subject":"Support customising nova-conductor and proxy groups for cells","status":"MERGED","created":"2019-12-13 17:31:04.000000000","updated":"2019-12-20 12:13:11.000000000","submitted":"2019-12-20 12:11:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698970-1576843905845-cc347a4a","meta_rev_id":"33c56f160a13f7afb754ae4d5eba9177e940616c","_number":698970,"virtual_id_number":698970,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 12:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-12-19 12:23:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-18 17:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-12-19 12:23:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-20 08:07:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 12:23:10.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-20 08:07:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-12-20 12:11:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b618e0ca20b530944fdf73299b6336b0a234151d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-13 17:31:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"beefe172cc5b46384799ad007e06b5314c22a64b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 19:47:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ffd9f95b3934c82b42d1f6cfde461ea : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6368fe45286f42b3be658b95e4511e00 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d19775c9f1d48909cc0229863e7729e : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b1fd261df84a71b8523c3e59f35ac6 : FAILURE in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae74d230199c4288b782b52d4d2de0b9 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e87324f444840e6a0154c298fc4b03a : FAILURE in 4m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f40206d562c4a799873e0d9a933c1b3 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44ce567eabe84e37a3ac8e4a55a97895 : FAILURE in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f32d49bab604795b2b03ab8694c7f3e : SUCCESS in 38m 28s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/be09bbcf43384876ae0a6ed3cf0342eb : SUCCESS in 1h 05m 23s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/446f10620b9a48259dafa392c3f3a18d : SUCCESS in 2h 13m 20s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/fbc4f2797e874ca1b534e63647da5249 : SUCCESS in 26m 49s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ae5130eb981943228777ae94acfbcd2a : SUCCESS in 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"62f57a4d618f27f1b3cc85d20ea1f0be7411e596","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-14 10:00:51.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nWe should have done this in kolla really.","accounts_in_message":[],"_revision_number":1},{"id":"31c3db4c5c997d9b5ac8fc3aa3c46b102042d771","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-16 09:25:02.000000000","message":"Patch Set 1:\n\nDo you mean something like creating a new group in Kolla-Ansible called `nova-cell-controller` and making it default to `nova`? Would we be able to backport that change in K-A to train?","accounts_in_message":[],"_revision_number":1},{"id":"f05c988944729ce6c462fb2f78d96769560ca364","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-16 10:26:27.000000000","message":"Patch Set 1:\n\n\u003e Do you mean something like creating a new group in Kolla-Ansible\n \u003e called `nova-cell-controller` and making it default to `nova`?\n \u003e Would we be able to backport that change in K-A to train?\n\nYes, or possibly \u0027cell-control\u0027 -\u003e \u0027nova-cell-control\u0027 -\u003e \u0027nova-conductor\u0027 to use the k-a 3 level model. I don\u0027t think we\u0027ll get it into train, but we could look at it for Ussuri. It could be made backwards compatible by linking cell-control to control.","accounts_in_message":[],"_revision_number":1},{"id":"32146e55c82fcfcd932b0ff7628c66eb64a0f742","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 15:53:17.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"152c7e6db7b3f379881e874e31e11785fd4e450e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 15:57:35.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"796fc3459b5371b86a3c3d4231b949fcb7a06e53","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 15:57:58.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"ea3f2ee75c61c9b7b6ab0cb484717cd1020741b2","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 17:01:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c3eb64006dc7d093f267638bb73e8d4d67ede4df","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 17:12:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e97e350cf0554c07e5549c0267e7a244b1d8dfe","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 17:31:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7d15bea9e841a629f04f562d53cd77b37a4b561b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 17:38:15.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"b36bf6efdcaab021f12a85ab26177e75a31f0eb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 19:39:29.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32ed19b319464417ae5843f934847fc2 : SUCCESS in 7m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/36601689c61e457b9eaf2d36740bd5a7 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40c7ae9574b94101b9b2af91fadf907c : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3231ecf63024f2aacd63b991ca419d6 : SUCCESS in 5m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14358b40e30641bbbb5911e02f52eddd : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef0fbc57d9a4e93a7f297a0f9eaec9a : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/375e5e36457f4ae6875ac3f5b4baf665 : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1b1c71c55284277a8f58bd67347a638 : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29086f9517124b059180e4845a1e79a0 : SUCCESS in 30m 59s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/3da42678e8d8409ab6be6bd6e155e226 : SUCCESS in 1h 16m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/825e7d1120c54709a40397b8f07bd235 : SUCCESS in 2h 04m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ba0ccf8b34e14471b8a18706b8513969 : SUCCESS in 22m 56s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7fb5faa1800a4a65acaa4667a3f2d805 : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":5},{"id":"d8246ed4b75879f9455e74895e25327e45d4f59d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 17:37:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7de82e95c3347129fdf2ef3bc8147f9a197f6f37","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 17:40:31.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"b8ede9554f5f16bfca70505ca0f0a5058761d8fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 19:33:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2aaa0ea8b8b4be8995f245c4a0e8940 : SUCCESS in 5m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/616926419c81435a84772f30101cd0d1 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ff2886bd0c43b9be7360fc2581c865 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfb6cd968aa942f5993d5ab6fd85db86 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a506e80b6f294f41968e01beccd87ac0 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc13bb7d54d24f12bf0590cbdebad8ca : SUCCESS in 7m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2517efcb6baf4643b5e949124952f4cb : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d7f1d879586847beb88b3049f84d0c82 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afc892bee7f349009473fd33223fa76d : SUCCESS in 35m 52s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8cd476663c8b4b81820198ac767a85d0 : SUCCESS in 1h 14m 37s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5546d9d45f344861900981985b77f840 : SUCCESS in 1h 51m 05s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5528258db07143339a319d411cca57de : FAILURE in 24m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/685331f753704c8d83c05953e6940043 : FAILURE in 39m 33s","accounts_in_message":[],"_revision_number":6},{"id":"025f6b604931698c04bd684e1ef57e60ee8291e1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-19 12:23:10.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"ce3bbafb2277c81f27203457e92e3a281c1c9a0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 14:11:43.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ea2a1e5fd6e4255bf8f5f3021b53dd4 : SUCCESS in 7m 23s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/632f0ade448a4da5813c73deb710726f : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97b89aa6733f48b582b8be3d0fde7629 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b16e0397a58428c87e02d37d3ebe148 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cc0e1e62b6044d2cbba0712a3dba5ea2 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ae93a2c5584b3e8a83dca4450cca2f : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ccba8e943454e6d856e3612b808f9e3 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/10220ecd27fd487ba4e33a72e979c70a : SUCCESS in 9m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6404163293c44412a5c599f0a33493a7 : SUCCESS in 30m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/989e360cb7334541b9acb38618cedcdb : SUCCESS in 1h 08m 49s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/780c4ac3d2b84e2fb1ac4f8297d3b4b7 : FAILURE in 1h 45m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/318a8ff8239f4175929b3ca8a367037b : FAILURE in 20m 29s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/42be6645797641f5a12c3aa54f99af09 : FAILURE in 37m 14s","accounts_in_message":[],"_revision_number":6},{"id":"2c22119f35e9994bfdb48a222712cd263c1fb060","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-19 14:12:54.000000000","message":"Patch Set 6:\n\nWe need to wait until the new Bifrost image is on Docker Hub.","accounts_in_message":[],"_revision_number":6},{"id":"1017925d1819a7dfd77a7e30d323476b2de6c39e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-20 08:07:51.000000000","message":"Patch Set 6: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"f2dcaffceb6e0ebfe0f1c7f0be7514439b9c23e5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-20 08:07:58.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"41b88dc415a435d992a1f3b2d64e2ce9e90e45c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:29:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2645dfd740449bdb2c3d37db107a30e : SUCCESS in 4m 07s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8449c625ed624fc58d0295dbfb4997b3 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54676cf14e94459a86be1b36fbc8cd22 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5020bf7946a249509cad4124a7092a69 : SUCCESS in 4m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/552101ee0fb34b93beb6d6a34135b82a : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5339c0fd6744868af544f0af2114493 : SUCCESS in 8m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cef6075ced1c4984b1646db75c328e1b : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a2127f5b9e32484d962db5da31667cbd : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d46e65088724796919c02eed1c842b6 : SUCCESS in 48m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e94bfab7f36745cbbde24b3fe6cef952 : SUCCESS in 1h 27m 31s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a70a302a894f4a00b4db5cae313a76ab : SUCCESS in 1h 59m 28s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c90cbe4700dd48039a9f3a9472a40be4 : SUCCESS in 22m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b825367300694335a4b3239daabe39e9 : SUCCESS in 34m 01s","accounts_in_message":[],"_revision_number":6},{"id":"1d88bae34e4cd8d0a3bcab17652095269b7b4037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 10:29:20.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3e8e6ae7e81ceb3daaf60e39c5431b2b04b7dca0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 12:11:45.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2a4ed9eeff24a8cbc574b2a667e115c : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0eec69f2739548208aedcea6035ac4a8 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1839950b3eee41ffb2707381a7439e71 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6d817266e53f411688c68d188a851284 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb8b6839a1fc4f429b46101b98353fce : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/637f34014de84d2c99d7fe4fb6d95500 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f26ee1e2602945d4a618f937d6f33ba7 : SUCCESS in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/30eefc1c69b344aaaa8ab8416fef3158 : SUCCESS in 30m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/606a23195f1f4e8f888e6398b2342416 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b136c70028224a448b4e0649f106fbdd : SUCCESS in 1h 41m 37s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e20c00411c944891b649e50ad75b8267 : SUCCESS in 26m 50s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d86693ada02140ecbacdc803f324dfa0 : SUCCESS in 41m 19s","accounts_in_message":[],"_revision_number":6},{"id":"19342b882b91b642564d49a66bbf760473568d74","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 12:11:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"d717ffd959178a31dfb5b1dd2fd32f1bfcb00637","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 12:13:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/149d98fa86084fe59fd88ffd81ed6bc8 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"05ae35bf6c06aceb6a4394152a48b333cbb44f13","revisions":{"199e4d91d07ff1c0fd9f196f84e5d50901edce66":{"kind":"REWORK","_number":1,"created":"2019-12-13 17:31:04.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/70/698970/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/698970/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/698970/1"}}},"commit":{"parents":[{"commit":"56b355c337be7f1913283f1e62a7e90fbc95d047","subject":"Support custom Kolla group_vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56b355c337be7f1913283f1e62a7e90fbc95d047"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 17:27:35.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 17:29:18.000000000","tz":0},"subject":"Support customising nova-conductor group for cells","message":"Support customising nova-conductor group for cells\n\nStory: 2004291\nTask: 37804\nChange-Id: Ife90d839a5184a3b692c49e0411cd1ac6467f25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/199e4d91d07ff1c0fd9f196f84e5d50901edce66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/199e4d91d07ff1c0fd9f196f84e5d50901edce66"}]},"branch":"refs/heads/master"},"c9b2d290a3f10739827e1edd8d4ee90eaf937d96":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-12-17 15:53:17.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/70/698970/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/698970/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/698970/2"}}},"commit":{"parents":[{"commit":"164c93e8988ce5d8c8fb0e07aac5e063f9b81e2d","subject":"Support custom Kolla group_vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/164c93e8988ce5d8c8fb0e07aac5e063f9b81e2d"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 17:27:35.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 15:52:50.000000000","tz":0},"subject":"Support customising nova-conductor group for cells","message":"Support customising nova-conductor group for cells\n\nStory: 2004291\nTask: 37804\nChange-Id: Ife90d839a5184a3b692c49e0411cd1ac6467f25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9b2d290a3f10739827e1edd8d4ee90eaf937d96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9b2d290a3f10739827e1edd8d4ee90eaf937d96"}]},"branch":"refs/heads/master"},"0a73d579f60cd8a6bd235c8f55ae0e587e8a2683":{"kind":"NO_CHANGE","_number":3,"created":"2019-12-17 15:57:35.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/70/698970/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/698970/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/698970/3"}}},"commit":{"parents":[{"commit":"1b85c53708605fea9c3e3a6562930ecb02011364","subject":"Support custom Kolla group_vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b85c53708605fea9c3e3a6562930ecb02011364"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 17:27:35.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 15:57:19.000000000","tz":0},"subject":"Support customising nova-conductor group for cells","message":"Support customising nova-conductor group for cells\n\nStory: 2004291\nTask: 37804\nChange-Id: Ife90d839a5184a3b692c49e0411cd1ac6467f25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a73d579f60cd8a6bd235c8f55ae0e587e8a2683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a73d579f60cd8a6bd235c8f55ae0e587e8a2683"}]},"branch":"refs/heads/master"},"711f85b3b4c09811db1e2f9fb1d16b35849c24e1":{"kind":"REWORK","_number":4,"created":"2019-12-17 17:01:51.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/70/698970/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/698970/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/698970/4"}}},"commit":{"parents":[{"commit":"ac7e17f3b22b81f3f943c48b5b5e9a3e394b083b","subject":"Support custom Kolla group_vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac7e17f3b22b81f3f943c48b5b5e9a3e394b083b"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 17:27:35.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 17:01:21.000000000","tz":0},"subject":"Support customising nova-conductor group for cells","message":"Support customising nova-conductor group for cells\n\nStory: 2004291\nTask: 37804\nChange-Id: Ife90d839a5184a3b692c49e0411cd1ac6467f25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/711f85b3b4c09811db1e2f9fb1d16b35849c24e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/711f85b3b4c09811db1e2f9fb1d16b35849c24e1"}]},"branch":"refs/heads/master"},"9216090ffdc399014f64e200ff5d40f909f32d8a":{"kind":"REWORK","_number":5,"created":"2019-12-17 17:31:59.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/70/698970/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/698970/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/698970/5"}}},"commit":{"parents":[{"commit":"ac7e17f3b22b81f3f943c48b5b5e9a3e394b083b","subject":"Support custom Kolla group_vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ac7e17f3b22b81f3f943c48b5b5e9a3e394b083b"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 17:27:35.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 17:30:56.000000000","tz":0},"subject":"Support customising nova-conductor and proxy groups for cells","message":"Support customising nova-conductor and proxy groups for cells\n\nStory: 2004291\nTask: 37804\nChange-Id: Ife90d839a5184a3b692c49e0411cd1ac6467f25a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9216090ffdc399014f64e200ff5d40f909f32d8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9216090ffdc399014f64e200ff5d40f909f32d8a"}]},"branch":"refs/heads/master"},"05ae35bf6c06aceb6a4394152a48b333cbb44f13":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2019-12-18 17:37:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/70/698970/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/70/698970/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/70/698970/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/70/698970/6"}}},"commit":{"parents":[{"commit":"72d4d64609030201ccdc753b98091552638b03e5","subject":"Support custom Kolla group_vars","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/72d4d64609030201ccdc753b98091552638b03e5"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 17:27:35.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-18 17:34:31.000000000","tz":60},"subject":"Support customising nova-conductor and proxy groups for cells","message":"Support customising nova-conductor and proxy groups for cells\n\nChange-Id: Ife90d839a5184a3b692c49e0411cd1ac6467f25a\nStory: 2004291\nTask: 37804\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05ae35bf6c06aceb6a4394152a48b333cbb44f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05ae35bf6c06aceb6a4394152a48b333cbb44f13"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
