)]}'
{"id":"openstack%2Fkayobe~698971","triplet_id":"openstack%2Fkayobe~master~Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f","project":"openstack/kayobe","branch":"master","topic":"story/2004291","hashtags":[],"change_id":"Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f","subject":"Add Nova Cells documentation","status":"MERGED","created":"2019-12-13 17:31:04.000000000","updated":"2019-12-20 13:44:46.000000000","submitted":"2019-12-20 13:43:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698971-1576849394530-bbef5f2b","meta_rev_id":"d843d608b56889c182e706599f34d0f97d9f2347","_number":698971,"virtual_id_number":698971,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 13:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-12-19 12:23:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-18 17:40:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2019-12-19 12:23:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-20 08:08:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-19 12:23:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-20 08:08:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-12-20 13:43:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6577cfe296d0acf8baa79d9a7477c34881ab4f3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-13 17:31:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b78bc2faa40ea38b1ed0e1e6852d75f3f6e2a93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:57:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5af258b18cb436a9b08c0810f4638e3 : FAILURE in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dbe4ae58f244bbb89d8eefe7fee31f4 : FAILURE in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"10911bcdf1d617355f94823bbec8edf5031a3a5c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 15:53:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc1aecf4aec18709d89d4ee4c4fe2deaf0d98b62","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 15:57:35.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b8be33efe1866aadf98a8ea1b098f12e3a1002e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 16:15:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf8e55ca1b9428b87df2e6865386e78 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25607af209ec47f2b613cfe04728da42 : SUCCESS in 10m 09s","accounts_in_message":[],"_revision_number":3},{"id":"66219b056eff1ebfb4fd0b3f0f7ee5b9a9f87f20","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 16:28:22.000000000","message":"Patch Set 3: Code-Review+2\n\nMakes sense. I don\u0027t think it needs to be too polished yet.","accounts_in_message":[],"_revision_number":3},{"id":"4a4087a27e89f607803a2753cff30bc7cf269851","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 17:01:51.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"1c58912cf3569952226caf241f99c478bce8d13d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 17:09:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"03f8b7aa5fee247c6805121ff6de93ad209f2ce3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 17:31:59.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f6ca215f7f32864ca82edd89fe6fcefc7cfcbe8b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 17:33:01.000000000","message":"Patch Set 6: Workflow-1","accounts_in_message":[],"_revision_number":6},{"id":"b0571b519c4e8cb3e16d317f2acaf2f6fe04e65a","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-12-17 17:54:50.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"1d3a51e27f31b59ebf09f23ba5d3feea05e93139","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-17 18:00:28.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"b8ab4e1036dd20ec2a2674dd15b3361ac22fc822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 19:48:59.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0519c24db8b74a3993437ff84f64494c : SUCCESS in 8m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b5aa9c9fbfa54538a2ff493a393113f1 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f091979f40d406d892bbd8a55250a6a : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d830a220c6154b8399d30c0ba9b0c263 : SUCCESS in 7m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5505984e07bb41b9bf38ca0db57d2c36 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcf8997961bc4795b01dfd63651f64ae : SUCCESS in 7m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/31e40c9c0e7c4d8586ec5d8365b80156 : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/71a8ea0459da4ec09bd04c8bd82dcc18 : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63b1ce373fc5497f9427a56e2c5163e5 : SUCCESS in 40m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/be0e5480970346da9cc8a16afa033fbd : SUCCESS in 1h 10m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b4c4c7fa33ef4577bac998eb0a872a35 : SUCCESS in 1h 50m 36s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/937ee9b0c30f4e24905128e1abd56b5a : SUCCESS in 20m 42s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/da33145336484c37a680c684ecd454a2 : SUCCESS in 1h 22m 36s","accounts_in_message":[],"_revision_number":7},{"id":"89934b4075b3a25f834f1ed127c9c56ad4bb9bed","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 17:37:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9e439f4431aa459b4921d3265d38630e898be3f0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-18 17:40:23.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"c2de76155ebfa280edccf026632df45c24e6e269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 19:32:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f86282793134585bb413dfd328d6717 : SUCCESS in 4m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2b01bcb7c4914cd391c0f436b1ae0f37 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78098f5de6314c0cb7247cb161995f68 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/332c71664e1244c0a48765628d54f303 : SUCCESS in 3m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d338d9f4214b4cd6b2071fe2e1e1c437 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e0f0038aba4b1b988e2db8a3433a0f : SUCCESS in 5m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/85641a8225fe45e1b400a6b6010bb0a3 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1b797612a89047c3bb259c76b7bb7ecd : SUCCESS in 13m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/696b94acff4c458fb2492b81d1b3250b : SUCCESS in 37m 26s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/598550b6ff8e4818b771d048051d04eb : SUCCESS in 1h 18m 56s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8f4f8d32cade40b99b25ccb56332c95f : SUCCESS in 1h 49m 39s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/752e4a18b41f468fa1a91c9f29c1efed : FAILURE in 24m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/8d45b0f508974d42b9611bfdb2e51ad2 : FAILURE in 41m 30s","accounts_in_message":[],"_revision_number":8},{"id":"c442825157896dd9e6cd40706c6cb1bedc042865","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-19 12:23:44.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"d4bf072e7a8aa745986e037f5d9bf317fecef813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 14:12:44.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a76db6f47301439da170f215bf4fe502 : SUCCESS in 5m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8ac00c8980db4959a217b08b7d728238 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06402f8de40a4ebd89c3b576903febd9 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3912e1d113b48009be9fe5168794dff : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2e798a1d51a5463c9518ed53b21661e2 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba4ffcbf480c44eeba0f2d6319bd9d45 : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/886a9e3af19c4df785c57be07c687682 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/799f23e000034a2183c230e3ba1d5fdc : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/13f3efd3e78e4f7497e5ab26b589c5ab : SUCCESS in 30m 57s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/eb5b43f78d724f8a8055559b979ae193 : SUCCESS in 1h 06m 02s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/15ef3e969652439a80d3b8a0ec34729d : SUCCESS in 1h 46m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/95bded9bf96d421bae39d2733d4f662c : FAILURE in 24m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/1f3bcd7b513d4cd5b50ad4e9b2657a0f : FAILURE in 35m 48s","accounts_in_message":[],"_revision_number":8},{"id":"f89b613641add3cb70480139f7e882c04b230fbb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-19 14:13:01.000000000","message":"Patch Set 8:\n\nWe need to wait until the new Bifrost image is on Docker Hub.","accounts_in_message":[],"_revision_number":8},{"id":"6b9552386b9b37d55983443c84080d2aa6d85495","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-20 08:08:24.000000000","message":"Patch Set 8: Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"48484c0d6dfceb0150abeeb751891f082a0772da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 09:57:01.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83043df615204fc892bc9344bfa06eb3 : SUCCESS in 4m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/fe95fa4f092a4a98903885256996fdbd : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/370e8448869d436cbc99087516b1573c : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f6c1d5bb179402196f8571d12572534 : SUCCESS in 5m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f3f01d196fe94a2b8d70d4932584c36a : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4491cd19a6f14c12bd26d28391f8b5b1 : SUCCESS in 9m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d26d06c7aa94eb0ab11ce9693f9f08f : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93679a4c207a4a20a9882e663818e841 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/208f719d1bf94db9aa363c27d93b1384 : SUCCESS in 38m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9297fc7910024b13980583f886b9bdfc : FAILURE in 1h 16m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7663640b225c4e369c1b724806a5d2d9 : SUCCESS in 1h 44m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/797f611440644cddb38d69e2ba4304d8 : SUCCESS in 30m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7cb8cdd42e9140ccb9c5bd1a65b312a1 : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":8},{"id":"f980b9909b17187590a70990bbe4cc9868013cbc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-20 10:16:36.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"39b684152d6de4813366e2065e6e18b29a47f1ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 12:03:07.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c51500ae42045108060f8fb791a1b50 : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/344284cae21542b79e0d493def6d7cd2 : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86f8aee851bb469abb78cc79456f782f : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd44fc9ca6144b38b300a48da0cfbe64 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f6af9168956540f28ef3fa940d182826 : SUCCESS in 4m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca3ac8979ba4408ba7c2106cbf9d6800 : SUCCESS in 5m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9da8d0a2537446be956a68839d8641f0 : SUCCESS in 4m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ad53b56025d489d89a4099928323d48 : SUCCESS in 8m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d437e2ba672541d186f549f9ed0f2827 : SUCCESS in 37m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d00db982d70f4f09b0beb6fdc19fcbb2 : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d8c013c6b99c4a85905c1b73debdc035 : SUCCESS in 1h 41m 53s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ce4e0300e8614cc28204aedff8cf48cb : SUCCESS in 20m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/686d113503ae4ee9803848241e24bdac : SUCCESS in 39m 05s","accounts_in_message":[],"_revision_number":8},{"id":"2806e599949ab731f275e0e072ec1856c57f6237","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 12:03:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"f2aaef408d692fd1dcd44670cdea6057a4139e42","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 13:43:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"e6d2993aaf3d847c64234bc94fb79287c37207d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 13:43:14.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/52280f6f90814a82bd42c11148e346f5 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95d89dac81de446fa7532792696c54ad : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/759c9767e4d3420eaef63b4e646f4ad9 : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/45b5ca535dfd41c0a88723c093e370cf : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46c6f606628a4f55863215a4f800e29f : SUCCESS in 5m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5344a631b33f4594ae1eb3d02c397285 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11551533688e48d389f9d17d6b6e80e7 : SUCCESS in 8m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4fdda0feb8044fccbb244422c171871b : SUCCESS in 30m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/648e9a3d5f2d4e488358e1d3634a71f0 : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/9ddfac1fbef3445e91eed7c11484ec9c : SUCCESS in 1h 38m 29s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/79aa90494c3c4bc48f2890ecc866c0ee : SUCCESS in 18m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/16e986b8cf214534bdcadf94726b27d2 : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":8},{"id":"e0d728b6d73883901ec53261fa30dd366ca5c37e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 13:44:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/674f7cb843154179b908723e09a7e957 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"aee2af7a758c92f50e0fddfff518f3d5e80aa9e5","revisions":{"7f10a5fb18219d819fd7a33caf9fee2d9de6a58b":{"kind":"REWORK","_number":1,"created":"2019-12-13 17:31:04.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/698971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/698971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/698971/1"}}},"commit":{"parents":[{"commit":"199e4d91d07ff1c0fd9f196f84e5d50901edce66","subject":"Support customising nova-conductor group for cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/199e4d91d07ff1c0fd9f196f84e5d50901edce66"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 16:03:59.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 17:29:30.000000000","tz":0},"subject":"WIP: Cells documentation","message":"WIP: Cells documentation\n\nClearly not very user friendly at the moment. We probably want to\ninclude these files as part of test.\n\nStory: 2004291\nTask: 37810\nChange-Id: Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f10a5fb18219d819fd7a33caf9fee2d9de6a58b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f10a5fb18219d819fd7a33caf9fee2d9de6a58b"}]},"branch":"refs/heads/master"},"94eb731db9860f1e91827830f3e29eb8e3ec9488":{"kind":"REWORK","_number":2,"created":"2019-12-17 15:53:17.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/698971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/698971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/698971/2"}}},"commit":{"parents":[{"commit":"c9b2d290a3f10739827e1edd8d4ee90eaf937d96","subject":"Support customising nova-conductor group for cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9b2d290a3f10739827e1edd8d4ee90eaf937d96"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 16:03:59.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 15:52:50.000000000","tz":0},"subject":"Cells documentation","message":"Cells documentation\n\nClearly not very user friendly at the moment. We probably want to\ninclude these files as part of test.\n\nStory: 2004291\nTask: 37810\nChange-Id: Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94eb731db9860f1e91827830f3e29eb8e3ec9488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94eb731db9860f1e91827830f3e29eb8e3ec9488"}]},"branch":"refs/heads/master"},"246a45405dc31c84e0363cacb8a57c0cb324e7a3":{"kind":"NO_CHANGE","_number":3,"created":"2019-12-17 15:57:35.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/698971/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/698971/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/698971/3"}}},"commit":{"parents":[{"commit":"0a73d579f60cd8a6bd235c8f55ae0e587e8a2683","subject":"Support customising nova-conductor group for cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0a73d579f60cd8a6bd235c8f55ae0e587e8a2683"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 16:03:59.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 15:57:20.000000000","tz":0},"subject":"Cells documentation","message":"Cells documentation\n\nClearly not very user friendly at the moment. We probably want to\ninclude these files as part of test.\n\nStory: 2004291\nTask: 37810\nChange-Id: Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/246a45405dc31c84e0363cacb8a57c0cb324e7a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/246a45405dc31c84e0363cacb8a57c0cb324e7a3"}]},"branch":"refs/heads/master"},"d18755a9818a1c847aaa466537d66ba4ad1e06f9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-17 17:01:51.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/698971/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/698971/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/698971/4"}}},"commit":{"parents":[{"commit":"711f85b3b4c09811db1e2f9fb1d16b35849c24e1","subject":"Support customising nova-conductor group for cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/711f85b3b4c09811db1e2f9fb1d16b35849c24e1"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 16:03:59.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 17:01:21.000000000","tz":0},"subject":"Cells documentation","message":"Cells documentation\n\nClearly not very user friendly at the moment. We probably want to\ninclude these files as part of test.\n\nStory: 2004291\nTask: 37810\nChange-Id: Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d18755a9818a1c847aaa466537d66ba4ad1e06f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d18755a9818a1c847aaa466537d66ba4ad1e06f9"}]},"branch":"refs/heads/master"},"7e85845108ea7338d61c7b2adbec5514ec8200c7":{"kind":"REWORK","_number":5,"created":"2019-12-17 17:09:26.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/698971/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/698971/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/698971/5"}}},"commit":{"parents":[{"commit":"711f85b3b4c09811db1e2f9fb1d16b35849c24e1","subject":"Support customising nova-conductor group for cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/711f85b3b4c09811db1e2f9fb1d16b35849c24e1"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 16:03:59.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 17:08:55.000000000","tz":0},"subject":"Cells documentation","message":"Cells documentation\n\nClearly not very user friendly at the moment. We probably want to\ninclude these files as part of test.\n\nStory: 2004291\nTask: 37810\nChange-Id: Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e85845108ea7338d61c7b2adbec5514ec8200c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e85845108ea7338d61c7b2adbec5514ec8200c7"}]},"branch":"refs/heads/master"},"0c3a0bbe8a63a4b3edc802f59d7d89c8f9cc8834":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-12-17 17:31:59.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/698971/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/698971/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/698971/6"}}},"commit":{"parents":[{"commit":"9216090ffdc399014f64e200ff5d40f909f32d8a","subject":"Support customising nova-conductor and proxy groups for cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9216090ffdc399014f64e200ff5d40f909f32d8a"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 16:03:59.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 17:31:20.000000000","tz":0},"subject":"Cells documentation","message":"Cells documentation\n\nClearly not very user friendly at the moment. We probably want to\ninclude these files as part of test.\n\nStory: 2004291\nTask: 37810\nChange-Id: Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c3a0bbe8a63a4b3edc802f59d7d89c8f9cc8834"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c3a0bbe8a63a4b3edc802f59d7d89c8f9cc8834"}]},"branch":"refs/heads/master"},"9999ec08ddc3d8ee9c1d4c19a3b403804e3af803":{"kind":"REWORK","_number":7,"created":"2019-12-17 17:54:50.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/71/698971/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/698971/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/698971/7"}}},"commit":{"parents":[{"commit":"9216090ffdc399014f64e200ff5d40f909f32d8a","subject":"Support customising nova-conductor and proxy groups for cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9216090ffdc399014f64e200ff5d40f909f32d8a"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 16:03:59.000000000","tz":0},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-17 17:54:30.000000000","tz":0},"subject":"Cells documentation","message":"Cells documentation\n\nClearly not very user friendly at the moment. We probably want to\ninclude these files as part of test.\n\nStory: 2004291\nTask: 37810\nChange-Id: Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9999ec08ddc3d8ee9c1d4c19a3b403804e3af803"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9999ec08ddc3d8ee9c1d4c19a3b403804e3af803"}]},"branch":"refs/heads/master"},"aee2af7a758c92f50e0fddfff518f3d5e80aa9e5":{"kind":"REWORK","_number":8,"created":"2019-12-18 17:37:31.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/698971/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/698971/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/698971/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/698971/8"}}},"commit":{"parents":[{"commit":"05ae35bf6c06aceb6a4394152a48b333cbb44f13","subject":"Support customising nova-conductor and proxy groups for cells","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/05ae35bf6c06aceb6a4394152a48b333cbb44f13"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2019-12-13 16:03:59.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2019-12-18 17:34:34.000000000","tz":60},"subject":"Add Nova Cells documentation","message":"Add Nova Cells documentation\n\nClearly not very user friendly at the moment. We probably want to\ninclude these files as part of test.\n\nChange-Id: Ib6ed0e68b55a98920cbaf3e8ec828901b4b9ed9f\nStory: 2004291\nTask: 37810\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aee2af7a758c92f50e0fddfff518f3d5e80aa9e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aee2af7a758c92f50e0fddfff518f3d5e80aa9e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
