)]}'
{"id":"openstack%2Fkayobe~700203","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ie6eea2c5d99f07cc9a4ca7eba5bbb0a87e369bb7","project":"openstack/kayobe","branch":"stable/train","topic":"create-train","hashtags":[],"change_id":"Ie6eea2c5d99f07cc9a4ca7eba5bbb0a87e369bb7","subject":"Update .gitreview for stable/train","status":"MERGED","created":"2019-12-20 16:49:21.000000000","updated":"2019-12-20 21:05:53.000000000","submitted":"2019-12-20 21:04:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700203-1576875869605-0e1c537b","meta_rev_id":"96d89fb716dac681f1be0aee86f8e150649c02fc","_number":700203,"virtual_id_number":700203,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-20 21:04:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-20 16:54:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2019-12-20 19:25:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 16:54:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2019-12-20 19:25:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2019-12-20 21:04:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a2e0a4e0bcbae99316732a452839011b4f6b34f0","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2019-12-20 16:49:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7444af2df9efdda098abebddbe4283abe8fde4f9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2019-12-20 16:54:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"622c517691b26cc03972c03783b9ab928efc4b0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 18:49:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a5259e9f34d48dd85050f0a0f740d51 : SUCCESS in 5m 13s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f8b801567fd04faebd7b107b8c84b564 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb5242071e6342f980d9a18b62fc7653 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41927dee9a8a4631b2cea2d895ec5c5b : SUCCESS in 3m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a77c5a147d5e4052810b674eb07da271 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f148234bacf642abb8eedb5800ba35bf : SUCCESS in 8m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b1a3e7e1ad14d6883775f82f16021be : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/66ad72ea89e946deb7b7d4bc5b7aafcc : SUCCESS in 8m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2b102832585d4971a71991561405bf53 : SUCCESS in 31m 12s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ee521a3e6bb14cb49f67e13145dc9414 : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a362182827ed4c16bd7f1cf2a686258a : SUCCESS in 1h 53m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7871d85e76f9471daa5af7b26002c67d : SUCCESS in 21m 00s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4dd52bbbaca34d44977af08eb44c1a45 : SUCCESS in 42m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d96fc5161d383fa5841f01b6038956237cb7b6e9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-12-20 19:25:31.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3563ca817aec0e963e5438593fcf13bd7b381ad0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 19:25:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f14e52c771b62cd231c1ce04f8b2eed652d68328","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 21:04:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ae82e9980d6646ea8dd805d4cf1b7d4d : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc6aae6d051042078a31e0b56f4a917b : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533e4a2946174ebebef76f28298427c3 : SUCCESS in 3m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/eaf40c8bf649469194d66dcddb1ea0e8 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdfd8bd2ca394372b693f6f7ac7cec83 : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fb7fdb2e70eb47aaa4c0bca39778b7a2 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b9bd93c245bb407a9e3a97e86738b388 : SUCCESS in 7m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/245a63ab97bc43d186187be75e0c2105 : SUCCESS in 37m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/84acad6d29e8439eb4638815a2051fa0 : SUCCESS in 1h 18m 36s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/52151110ca8d44539f7105d051e6c066 : SUCCESS in 1h 38m 28s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/979ba0d0c7234cd68f9e428c38f542b7 : SUCCESS in 20m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/dd7b6d5584f248a4b4f5306b9f4e0486 : SUCCESS in 35m 21s","accounts_in_message":[],"_revision_number":1},{"id":"dad33ab9c7b7a1e87dca86ad0f2db956b66b7839","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 21:04:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e8a457b46f07c7c45c94bb25b2a0105d7dbb7534","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 21:05:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d84cde6d82444b884ca358230b60928 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51fb5c68833775224e6141476f2f155921cb0b2d","revisions":{"51fb5c68833775224e6141476f2f155921cb0b2d":{"kind":"REWORK","_number":1,"created":"2019-12-20 16:49:21.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/03/700203/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/03/700203/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/700203/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/700203/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/03/700203/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/03/700203/1"}}},"commit":{"parents":[{"commit":"47aeacbacb0fca4e2547b2e61196d65ed4182b11","subject":"Merge \"Add Nova Cells documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47aeacbacb0fca4e2547b2e61196d65ed4182b11"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-12-20 16:49:20.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2019-12-20 16:49:20.000000000","tz":0},"subject":"Update .gitreview for stable/train","message":"Update .gitreview for stable/train\n\nChange-Id: Ie6eea2c5d99f07cc9a4ca7eba5bbb0a87e369bb7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51fb5c68833775224e6141476f2f155921cb0b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51fb5c68833775224e6141476f2f155921cb0b2d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
