)]}'
{"id":"openstack%2Fkayobe~701031","triplet_id":"openstack%2Fkayobe~master~I70d20381bec3f5b92bb467e6c506e32bb13548c2","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I70d20381bec3f5b92bb467e6c506e32bb13548c2","subject":"Fix IPA builds without packages specified","status":"MERGED","created":"2020-01-03 14:47:02.000000000","updated":"2020-01-06 16:28:50.000000000","submitted":"2020-01-06 15:58:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701031-1578326312131-17819b5d","meta_rev_id":"0701be3da256456880308be2172fd3474d84d840","_number":701031,"virtual_id_number":701031,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 15:58:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-01-06 14:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-06 08:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-01-06 14:06:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-01-03 15:20:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 08:05:21.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-01-06 14:06:45.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-01-06 15:58:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"876e344f63d1f760f09bc10c9d90c66d8efb9410","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 14:47:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"24578ba471765e90411688dca90ea68cbce84dbb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 14:57:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f94cfa6ba7c8bb201a19208abb6c4668ce4d5a70","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 15:20:09.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"5a3b281cd621c7d24b1f1df3b270a38632fcdfe6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 15:20:42.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"d0f4401a71cc2bd9548acd6ea062da5a525d2fec","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 15:20:43.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"4355da2db910ca5ed24c1b4f71060de050254254","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 17:12:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c59a3d95d2224267908ffcee27501c81 : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/15283f75c1234a96b6e506f685312a34 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d32e9d9a9bcd46b1b4eb05559efab65e : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d66b2296a71b4ecb802265eb48523d92 : SUCCESS in 3m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/013e76f3f6f440fd8409dd32d18a29b9 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da37ec9820614d06a901ebe95161ba28 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f6a9cb0b93d4bd99f326454fd96be16 : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a11af5498b3c4a238e95a9f9c2a875bf : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/10efe522532c497d967c6b792c2be728 : SUCCESS in 8m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c87bacea2b146b0ba84dfb48ab4e92c : SUCCESS in 31m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2bb373326e4a4ea483361ed6bb05eeb8 : SUCCESS in 1h 07m 27s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/cdbcc56049204df7ad5ef290144399aa : SUCCESS in 1h 46m 43s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c53e55e9c601430ea4d4e948fdee49b0 : SUCCESS in 20m 07s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/0bbd17ed222a46c884cc1d5468eb25cd : SUCCESS in 35m 26s","accounts_in_message":[],"_revision_number":3},{"id":"29ce5e48b142f82f08e40c834bfd5347c947acae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-06 08:05:21.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks Mark.","accounts_in_message":[],"_revision_number":3},{"id":"1207bf18dda3215ead537cb5898b3e06182a4d2d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-06 14:06:45.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"70e046ca82896d3471899bb1c32da6bef2359fc4","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-06 14:06:53.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"afa431fa41193f47f7f74498e658dfd23030389c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 14:06:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8253ac870819aed724588276fb605be23691dc28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 15:58:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3c8f5261c9054a55a9911405652e9ca1 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/95ec2bfb7f9a4f77a5f44fcd7ac8a589 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/148dfd56a488477f8c773e16420f9328 : SUCCESS in 5m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/12becfed6e6e41929b21838c38b980db : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02fbd014908c42978ce9191c0a2e769c : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2fd7f12456a4f5c8276caf6e2034afa : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f15e45925a82455da933b0368534f32c : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75d7c95bccdd4676a784ab103c6a7cb4 : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ff53673b317412791a31f4a1bcf1881 : SUCCESS in 33m 12s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/238d05c6a98f4d35aefc9bb1128c6316 : SUCCESS in 1h 12m 51s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7e88f6e7c765459f93a69a241b2e3109 : SUCCESS in 1h 50m 11s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0f9454e493d149eeac6e4551d822e6ae : SUCCESS in 22m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/b3c3aba620b44892b2d31ccb46d8da53 : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":3},{"id":"d8d803ddd7ff48d245d17c004fbe0b16c6cf8b0c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 15:58:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"394ce52c04a484e3759659ee019885426ed7aa35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:01:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6e78b518b274e30a49ee535cb9bf78d : SUCCESS in 2m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/405b2ce8212e4a03b456bcae2a21e7e1 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":3},{"id":"0701be3da256456880308be2172fd3474d84d840","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-06 16:28:50.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 3df96433f776457d99502b8b80caeff201189bd6","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"beba4d29b56df71b4d5c2bf881b663f86db6b752","revisions":{"616b5668294369ef94bcaa21193c676e5b5435bd":{"kind":"REWORK","_number":1,"created":"2020-01-03 14:47:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/701031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/701031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/701031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/701031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/701031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/701031/1"}}},"commit":{"parents":[{"commit":"4acc1b64ecce062db583f4e27a9e143d48b9752c","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4acc1b64ecce062db583f4e27a9e143d48b9752c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 14:46:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 14:46:03.000000000","tz":0},"subject":"Fix IPA builds without packages specified","message":"Fix IPA builds without packages specified\n\nSince story 2006855, it is possible to specify a list of packages when\nbuilding IPA images. However, this introduced an issue in the case where\nno packages are specified. See\nhttps://github.com/stackhpc/ansible-role-os-images/issues/26.\n\nThis change fixes the issue by updating stackhpc.os-images role to\nv1.7.0, which has a fix for this issue.\n\nChange-Id: I70d20381bec3f5b92bb467e6c506e32bb13548c2\nStory: 2007069\nTask: 37952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/616b5668294369ef94bcaa21193c676e5b5435bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/616b5668294369ef94bcaa21193c676e5b5435bd"}]},"branch":"refs/heads/master"},"cbccd3a04fcda0e8cc52807038deb1507c2f96a7":{"kind":"REWORK","_number":2,"created":"2020-01-03 14:57:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/701031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/701031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/701031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/701031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/701031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/701031/2"}}},"commit":{"parents":[{"commit":"4acc1b64ecce062db583f4e27a9e143d48b9752c","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4acc1b64ecce062db583f4e27a9e143d48b9752c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 14:46:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 14:57:39.000000000","tz":0},"subject":"Fix IPA builds without packages specified","message":"Fix IPA builds without packages specified\n\nSince story 2006855, it is possible to specify a list of packages when\nbuilding IPA images. However, this introduced an issue in the case where\nno packages are specified. See\nhttps://github.com/stackhpc/ansible-role-os-images/issues/26.\n\nThis change fixes the issue by updating stackhpc.os-images role to\nv1.7.0, which has a fix for this issue.\n\nChange-Id: I70d20381bec3f5b92bb467e6c506e32bb13548c2\nStory: 2007069\nTask: 37952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbccd3a04fcda0e8cc52807038deb1507c2f96a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbccd3a04fcda0e8cc52807038deb1507c2f96a7"}]},"branch":"refs/heads/master"},"beba4d29b56df71b4d5c2bf881b663f86db6b752":{"kind":"NO_CHANGE","_number":3,"created":"2020-01-03 15:20:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/701031/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/701031/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/701031/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/701031/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/701031/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/701031/3"}}},"commit":{"parents":[{"commit":"4acc1b64ecce062db583f4e27a9e143d48b9752c","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4acc1b64ecce062db583f4e27a9e143d48b9752c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 14:46:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 15:20:36.000000000","tz":0},"subject":"Fix IPA builds without packages specified","message":"Fix IPA builds without packages specified\n\nSince story 2006855, it is possible to specify a list of packages when\nbuilding IPA images. However, this introduced an issue in the case where\nno packages are specified. See\nhttps://github.com/stackhpc/ansible-role-os-images/issues/26.\n\nThis change fixes the issue by updating stackhpc.os-images role to\nv1.7.0, which has a fix for this issue.\n\nChange-Id: I70d20381bec3f5b92bb467e6c506e32bb13548c2\nStory: 2007069\nTask: 37952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/beba4d29b56df71b4d5c2bf881b663f86db6b752"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/beba4d29b56df71b4d5c2bf881b663f86db6b752"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
