)]}'
{"id":"openstack%2Fkayobe~701032","triplet_id":"openstack%2Fkayobe~master~I7f75c25602fd7ae4bfeb6abbdd3b42d8ee465abf","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I7f75c25602fd7ae4bfeb6abbdd3b42d8ee465abf","subject":"Fix overcloud provisioning due to Bifrost IPA rename","status":"MERGED","created":"2020-01-03 15:17:04.000000000","updated":"2020-01-06 16:28:31.000000000","submitted":"2020-01-06 16:02:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701032-1578326571558-a981c286","meta_rev_id":"aed2879b9f35c5e67d5c85e3f0b700475f54f128","_number":701032,"virtual_id_number":701032,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 16:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-01-06 14:15:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-06 08:11:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-01-06 14:15:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-01-03 15:18:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 08:11:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-01-06 14:15:08.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-01-06 16:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2e9fb6c450644487e5cbf3a87ca1d456a0576f01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 15:17:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfebb56f95a704693a5035879e1af8af3a60394a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-03 15:18:48.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"d657e757d1899cefac16a97b6b9d47c96e59438d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-03 17:17:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b389057071c4967a602738aa6a60367 : SUCCESS in 3m 46s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/145c2bc17cfe4603872c14ffb60fdf2a : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5ad9c2015c34d8799cd57d502ed54b9 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6279010415fb4f09b9f39989ebd2b8d3 : SUCCESS in 3m 49s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f9f48cf0aa9c4a299d0ee45b171f7873 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efd5eff6783d4fa48824488eacbf061b : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/174247fa67df441d80579a43bd55bc51 : SUCCESS in 2m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c263a7f7962481b96e8d360c7b28dc0 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/17ae5b0a28354aca8e46695dae87beb8 : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cf82f0447ef34db6a95a1ddafc36fab6 : SUCCESS in 37m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bd804302b44a48babdf36fc83b25ffa3 : SUCCESS in 1h 03m 56s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f3745e0be7124e2b97b330d2411b8104 : SUCCESS in 1h 54m 57s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/47db2aa9314e4bc8b836f88acbd3742a : SUCCESS in 20m 16s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3ceff86b777c4194aecfa24467bc8848 : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ef3c0f0bdd1b3a2dced1c7b4f70fb7fefbe9c76e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-06 08:11:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6ba19d45b6769e4a8b275ed3abc0d086137f2dde","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-06 14:15:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSeems reasonable","accounts_in_message":[],"_revision_number":1},{"id":"8102dd37a4c4170d5a245577b5220f291a89fe4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 14:15:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f88f76aa499f2f2d14913a57922cf1820ec5c379","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:02:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e8ff9cb6989fe916e8091098ec43653d79f767c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:02:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/06d63825ee1f4789b17cac664309d045 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4531ebd78fb340bfba04be9797a31469 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62c48d05d59c4d11be92f38045530dad : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd4cf58881954e748d66f8a75dd42ad2 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d482966a743c46f5adacaa3adf5976fe : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e646d14911104dfe81aed8cc53ff9fc7 : SUCCESS in 3m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/630f5b0f5f8f4f2d982b4bc6338f28d9 : SUCCESS in 4m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8286fcc7b26847a693226abf349ac539 : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c574ca98c4a488c90d569bd361f7904 : SUCCESS in 40m 10s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/e7d77f3d2417464586834f806788a6de : SUCCESS in 1h 06m 24s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/f80c5dc2a0c147019753bcb6f06ab83d : SUCCESS in 1h 46m 37s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b0f7f3a1fade4c0b962e65f94f2e9589 : SUCCESS in 24m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d09d78ad0cc04faa89f0fe5429af7394 : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d5ae736f9d174ad93f4ac6f76150856cf737b291","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 16:05:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/980f04160f084e65b1afaf0f4a64c5b1 : SUCCESS in 1m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41667b5753764bad980e8e2d297b74c6 : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1},{"id":"aed2879b9f35c5e67d5c85e3f0b700475f54f128","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-06 16:28:31.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 2b0a68d25cfad247623acf1264fef3a27f892157","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fa7d29b47eb8f23411265cd4581e077d100d0dc7","revisions":{"fa7d29b47eb8f23411265cd4581e077d100d0dc7":{"kind":"REWORK","_number":1,"created":"2020-01-03 15:17:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/701032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/701032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/701032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/701032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/701032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/701032/1"}}},"commit":{"parents":[{"commit":"4acc1b64ecce062db583f4e27a9e143d48b9752c","subject":"Update master for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4acc1b64ecce062db583f4e27a9e143d48b9752c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 15:07:07.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 15:07:07.000000000","tz":0},"subject":"Fix overcloud provisioning due to Bifrost IPA rename","message":"Fix overcloud provisioning due to Bifrost IPA rename\n\nKolla recently upgraded bifrost from 7.0.0 to 7.1.0 on the stable/train\nbranch. This switched to IPA builder to build the IPA image, and\nintroduced a rename of the IPA kernel file from ipa.vmlinuz to\nipa.kernel, which breaks overcloud provisioning. The iPXE kernel\ndownload fails with a 404, since Kayobe introspection rules use\nipa.vmlinuz for the driver_info.deploy_kernel URL.\n\nThis change works around the issue by setting two Bifrost variables,\nipa_kernel and ipa_kernel_url, to reference the old kernel filename of\nipa.vmlinuz. This works both in the case where the image is downloaded\nfrom a URL (ipa_kernel sets the destination file name), and where it is\nbuilt via \u0027kayobe seed deployment image build\u0027 (kayobe uses the legacy\nironic-agent DIB element rather than IPA builder, which creates a\nhardlink to ipa.vmlinuz).\n\nWe chose the above approach rather than switching to IPA builder due to\nit being a less risky change at a time close to release. A future\nrelease of Kayobe should switch to IPA builder, but this will be a\nlarger effort.\n\n[1] https://review.opendev.org/#/c/692200/1/playbooks/roles/bifrost-ironic-install/defaults/main.yml\n\nChange-Id: I7f75c25602fd7ae4bfeb6abbdd3b42d8ee465abf\nStory: 2007068\nTask: 37951\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa7d29b47eb8f23411265cd4581e077d100d0dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa7d29b47eb8f23411265cd4581e077d100d0dc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
