)]}'
{"id":"openstack%2Fkayobe~701245","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I7f75c25602fd7ae4bfeb6abbdd3b42d8ee465abf","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I7f75c25602fd7ae4bfeb6abbdd3b42d8ee465abf","subject":"Fix overcloud provisioning due to Bifrost IPA rename","status":"MERGED","created":"2020-01-06 16:28:31.000000000","updated":"2020-01-06 20:09:07.000000000","submitted":"2020-01-06 20:07:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701245-1578341236143-7a9322a4","meta_rev_id":"7ea1fe936aac920a1c29327c681716592a527bed","_number":701245,"virtual_id_number":701245,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 20:07:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-01-06 16:42:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-06 16:54:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-06 16:54:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 16:42:07.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-01-06 16:54:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-01-06 20:07:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bca2c5866641600d8c9a276f1cad01d3f35813a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-06 16:28:31.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fd5fc3224aac8d8a671c7d37760d7f05090682d4","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-06 16:42:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7a513e0cce1052ca1ae1ae5394262913bad158c2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-06 16:54:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ba589b6612c2bf6ee2603468b195e141dd7db437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 18:15:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6ee2517d0522490187babc36676fbdca : SUCCESS in 5m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b2baf506fe76403da01e6fc10d2d62c3 : SUCCESS in 7m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a806d0b8dc1c43ed9da13e7c79b5ad41 : SUCCESS in 8m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11c81821a887423298efced8f0ac7098 : SUCCESS in 5m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/26571778640540b59e2ab1a5e9b63da5 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198977718e054a0d9857b9c0f0f57bb4 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46dea16d6ed94f4db4b516f3fc6c5a6f : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1e44c387c3e44b889ba9f8085a833053 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f72e99c85bb04a41839d255a9fac460d : SUCCESS in 10m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ba8a35d5f7244f078996de6a422108e6 : SUCCESS in 32m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7c25d8eea2a0476ab21041e7232fe07f : SUCCESS in 1h 10m 53s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7d4628d8c1be4090bd076614284b3f6d : SUCCESS in 1h 45m 25s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/491496c7214845adaa5801afd0e06288 : SUCCESS in 21m 19s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/91edfe2d5292493790bd69db6e74bd85 : SUCCESS in 42m 16s","accounts_in_message":[],"_revision_number":1},{"id":"5180e864271fb6858900e78ef6a57795d777d342","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 18:15:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"80cd18a5170530b4cad7fef3c51b79847accde03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 20:07:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bd3629c10d84453aa0091cd0820ff7d4 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56450e3776834bdd83aaa60d541d5f11 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9313cbb45e1a4469b4cd728cf341ec8d : SUCCESS in 6m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/250dcacea348426d929c01b986cde5f2 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca3339cc2e34e1ba653d49509c46367 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcc746646be345b0b43b4217f15a84f0 : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/63a35ff17f9e45d4b55bd4ae5d6a173d : SUCCESS in 7m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a86bc67fb604d1c8e9c8421b55ea208 : SUCCESS in 9m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d34513a9eb741f3915a498cfb7532af : SUCCESS in 32m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/bfbbd86a09044cc0ae1d9e3bccbe6b80 : SUCCESS in 1h 17m 37s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5e13db574b734cbcb3f18bc4b07eeeca : SUCCESS in 1h 50m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3e77dc9afd74430b8dd58d8aa4938797 : SUCCESS in 23m 19s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/6daeaee6ff35494598e5c45050a7829a : SUCCESS in 36m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bbf34471f9223a519fb56b4d984602b62b63bbd7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 20:07:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7ea1fe936aac920a1c29327c681716592a527bed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 20:09:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985e014484b94845bc1c258c10237510 : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/510638b4cec54ea1b2ef4feaa3ee2fb3 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b0a68d25cfad247623acf1264fef3a27f892157","revisions":{"2b0a68d25cfad247623acf1264fef3a27f892157":{"kind":"REWORK","_number":1,"created":"2020-01-06 16:28:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/45/701245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/45/701245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/701245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/701245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/45/701245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/45/701245/1"}}},"commit":{"parents":[{"commit":"51fb5c68833775224e6141476f2f155921cb0b2d","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51fb5c68833775224e6141476f2f155921cb0b2d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 15:07:07.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-06 16:28:31.000000000","tz":0},"subject":"Fix overcloud provisioning due to Bifrost IPA rename","message":"Fix overcloud provisioning due to Bifrost IPA rename\n\nKolla recently upgraded bifrost from 7.0.0 to 7.1.0 on the stable/train\nbranch. This switched to IPA builder to build the IPA image, and\nintroduced a rename of the IPA kernel file from ipa.vmlinuz to\nipa.kernel, which breaks overcloud provisioning. The iPXE kernel\ndownload fails with a 404, since Kayobe introspection rules use\nipa.vmlinuz for the driver_info.deploy_kernel URL.\n\nThis change works around the issue by setting two Bifrost variables,\nipa_kernel and ipa_kernel_url, to reference the old kernel filename of\nipa.vmlinuz. This works both in the case where the image is downloaded\nfrom a URL (ipa_kernel sets the destination file name), and where it is\nbuilt via \u0027kayobe seed deployment image build\u0027 (kayobe uses the legacy\nironic-agent DIB element rather than IPA builder, which creates a\nhardlink to ipa.vmlinuz).\n\nWe chose the above approach rather than switching to IPA builder due to\nit being a less risky change at a time close to release. A future\nrelease of Kayobe should switch to IPA builder, but this will be a\nlarger effort.\n\n[1] https://review.opendev.org/#/c/692200/1/playbooks/roles/bifrost-ironic-install/defaults/main.yml\n\nChange-Id: I7f75c25602fd7ae4bfeb6abbdd3b42d8ee465abf\nStory: 2007068\nTask: 37951\n(cherry picked from commit fa7d29b47eb8f23411265cd4581e077d100d0dc7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b0a68d25cfad247623acf1264fef3a27f892157"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2b0a68d25cfad247623acf1264fef3a27f892157"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
