)]}'
{"id":"openstack%2Fkayobe~701246","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I70d20381bec3f5b92bb467e6c506e32bb13548c2","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I70d20381bec3f5b92bb467e6c506e32bb13548c2","subject":"Fix IPA builds without packages specified","status":"MERGED","created":"2020-01-06 16:28:50.000000000","updated":"2020-01-06 20:34:38.000000000","submitted":"2020-01-06 20:32:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701246-1578342777225-d3fc192b","meta_rev_id":"5eb72f10050225c313f571a7a41b99e4aa9c5337","_number":701246,"virtual_id_number":701246,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-06 20:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-01-06 16:41:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-06 16:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-06 16:53:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 16:41:44.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-01-06 16:53:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-01-06 20:32:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a253f7d138b64b55fc87e69178f11da718e7484","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-06 16:28:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a7abffc472d4c9fa1ee3d05735552323f82dd574","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-01-06 16:41:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f86f661d629fc6572b94801ac1616c54894cc885","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-06 16:53:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2dd322e53511ca3ff141d491e521311d3cd29360","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 18:47:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b21378976144490ad2c9eede0fe8ffa : SUCCESS in 6m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c8027d0239ab40c0bcddc61690e309cf : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/94269487352c4fec88bdf5de516a09f0 : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3416689aaa42fbaaa0371c49a4f257 : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59c1f8191e1b406ebe9452c7d7727317 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aba4c826b9b34abaaf591f8f521a7723 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0b353b7d622417a9ab0cf2960fc89ec : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f3aa5fbe9dfd453c9cf92ffa8641c17b : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6248fd62ccc24b858314f0eb70726ea5 : SUCCESS in 11m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e3c2316e90e4147846b3ad492bb7c9d : SUCCESS in 32m 22s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/025f4c6d9f194a288ae3c5d89ddc13ed : SUCCESS in 1h 12m 01s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/173e969f855d4dbc819097723faf4b9a : SUCCESS in 2h 13m 06s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/30f0126b089c44df9e59497e989ade34 : SUCCESS in 21m 13s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/bf835172d63749d6b86ffee4f304c62e : SUCCESS in 42m 16s","accounts_in_message":[],"_revision_number":1},{"id":"a9c333c7a5aaf752e948b50d5fe2b812de7c1c9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 18:47:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e2d9f61d9a9e8cc6c2808b64023a0eab7626685","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 20:32:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e41a636399024234bc0398781b97c245 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06380b818cbb41039c0bf18c679e76ec : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/670e241487eb405daebeb52d36355ddd : SUCCESS in 4m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/23b817c1e3fa4e84b486ca38342ca5e4 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2182cac2f6b1408a8c4467c68aa1e1b0 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8cfcf8da02564693a2fe523a1128d962 : SUCCESS in 3m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3509ba7a9d27486faecb6f0f9338d0dc : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/46dc6f648f5e4d41bd99e41e81c8d029 : SUCCESS in 9m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8010cc3aa3034addb385455efff0b6ad : SUCCESS in 30m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5f6eae6d427d46c5b13fec16916c0d64 : SUCCESS in 1h 07m 05s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a82f98f6e81d4e5b9d5c8ceab0ae83fb : SUCCESS in 1h 43m 51s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/97634166181641128aff692e0072fd0f : SUCCESS in 21m 35s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/2a4e4c8c11394406a2fe05c954c2d1a1 : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":1},{"id":"a90de47c1288f616faec836661f999a0e901b737","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 20:32:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5eb72f10050225c313f571a7a41b99e4aa9c5337","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-06 20:34:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82efb5dfe68f4edb9d4059a5e06bd5bd : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9904f53e149940b0ad9225afd1ab1f5c : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3df96433f776457d99502b8b80caeff201189bd6","revisions":{"3df96433f776457d99502b8b80caeff201189bd6":{"kind":"REWORK","_number":1,"created":"2020-01-06 16:28:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/701246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/701246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/701246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/701246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/701246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/701246/1"}}},"commit":{"parents":[{"commit":"51fb5c68833775224e6141476f2f155921cb0b2d","subject":"Update .gitreview for stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51fb5c68833775224e6141476f2f155921cb0b2d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-03 14:46:00.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-06 16:28:50.000000000","tz":0},"subject":"Fix IPA builds without packages specified","message":"Fix IPA builds without packages specified\n\nSince story 2006855, it is possible to specify a list of packages when\nbuilding IPA images. However, this introduced an issue in the case where\nno packages are specified. See\nhttps://github.com/stackhpc/ansible-role-os-images/issues/26.\n\nThis change fixes the issue by updating stackhpc.os-images role to\nv1.7.0, which has a fix for this issue.\n\nChange-Id: I70d20381bec3f5b92bb467e6c506e32bb13548c2\nStory: 2007069\nTask: 37952\n(cherry picked from commit beba4d29b56df71b4d5c2bf881b663f86db6b752)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3df96433f776457d99502b8b80caeff201189bd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3df96433f776457d99502b8b80caeff201189bd6"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
