)]}'
{"id":"openstack%2Fkayobe~703667","triplet_id":"openstack%2Fkayobe~master~I7c65689808cc0bf592a06b37f570db5b5cf7480c","project":"openstack/kayobe","branch":"master","attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-10-31 14:56:59.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},"8726":{"account":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"last_update":"2024-02-16 14:41:16.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-05-31 09:21:11.000000000","reason":"Vote got outdated and was removed: Code-Review+2"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-06-03 08:57:44.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}}},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-31 22:39:12.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-10-30 18:15:23.000000000","reason":"\u003cGERRIT_ACCOUNT_35263\u003e replied on the change","reason_account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}}},"hashtags":[],"change_id":"I7c65689808cc0bf592a06b37f570db5b5cf7480c","subject":"Add vagrant-libvirt support","status":"NEW","created":"2020-01-21 18:45:25.000000000","updated":"2026-02-19 15:52:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":31,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8e0ecdcbb4624e4ad614b5c7fb287797b3fc7e3b","_number":703667,"virtual_id_number":703667,"owner":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-31 17:09:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2024-10-31 22:39:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":2,"date":"2024-10-31 14:56:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2024-10-31 14:52:48.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"CC":[{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-21 20:33:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-27 11:21:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-02-07 15:34:30.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-02-16 14:41:16.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"CC"},{"updated":"2024-05-01 09:06:06.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2024-05-31 10:22:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-10-16 12:00:07.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2026-02-19 15:52:21.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"CC"}],"messages":[{"id":"608fb01011bbaa1b45671c346370383696d730ba","author":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"date":"2020-01-21 18:45:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7809400df43e9fd0b00badc921d7dee6a7d1ea6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 20:33:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7659626d3ac41a9ac612e4518d813e6 : SUCCESS in 5m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/aa5720353d7b4012adc91957e8d58168 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5749c984d4c8486ea3deed4a46869b8a : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d9c615313ef41e0b6dd5365b1274f06 : SUCCESS in 4m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ea9a962c11c486bae48b0092f0da559 : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6516dc22c944635b0ea52bfe0045a08 : SUCCESS in 7m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/20e595ab8d104a218391d8bde11724e3 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/250836028728418888afb177eda42348 : SUCCESS in 9m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1bb09376cd2b41c2b131aec20449adfc : SUCCESS in 38m 26s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c0e704ef8c5348e8bd91f8f30d82f6f0 : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5fe7c4e25775449cb08af56f185143e6 : SUCCESS in 1h 40m 23s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ff88a8cfc8a24e85ad2ec3375ae9bed2 : SUCCESS in 20m 57s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ba85469c7ab2427ebecb6998b108e1ac : SUCCESS in 36m 08s","accounts_in_message":[],"_revision_number":1},{"id":"9b10057e523c194d4bb6caebf1cb92d4d4eb0ca9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-27 11:21:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nThanks, could you also update the documentation? https://docs.openstack.org/kayobe/latest/development/vagrant.html","accounts_in_message":[],"_revision_number":1},{"id":"667b22dcd18e03cb82363a6f45f46e6f7c7b682e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-16 14:41:16.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"bec3142beecb121a2a7c6c5da0342fb0d04e4f9c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-02-16 15:00:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee42eeba969f309d025e5b4cc145e594edd28925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-20 15:07:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6e12b775a2996baf05639d2b9197adfbedebc47f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-20 15:14:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2487203a4f7e1f2fb04c239a6a60c206f9e97d28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-02-20 16:59:16.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"7e2c93f0153a99c7bfd54f05936962e7bfeb6905","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-20 19:01:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18871dc40db64243af4b2c291e66f4a1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd200d5a6ecd4626b330e9e4dc1438cd : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3760ff333567410fa10f9aff0d4f042b : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0444db6e76943dfa498d4590d058a2c : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/139e3e7f31204704afcb08e80300fce6 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c2c740700f547e79d17438e987e71cf : FAILURE in 11m 16s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c6361cef581849bc9d83871052de7e16 : SUCCESS in 9m 23s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/960ec393cc6942f3acced6888db2e66e : SUCCESS in 1h 03m 20s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/8e56fca48980479588960863a2829ef8 : SUCCESS in 56m 56s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d192c75a66174c759c1cc79bcb92c15c : SUCCESS in 55m 46s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d51c8347e3f14362ba61a98ead2c93e5 : SUCCESS in 1h 09m 57s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/ecd5b4fe819c4e00b68352679d7c7e56 : SUCCESS in 17m 25s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/d9595016211942b6810e74f0d553f70d : SUCCESS in 16m 23s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f2fb34083514563ae0333f816cb54ec : SUCCESS in 13m 04s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/c47d60cf10ee4f5e99e5e69094e07533 : SUCCESS in 1h 59m 53s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5408424c4be44cc489d1ec20bf248d8e : SUCCESS in 2h 00m 14s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/641bc81ad75549a3b850d3ff54e32e5b : SUCCESS in 18m 24s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/afa964b9a2fa46239430c819687f8290 : SUCCESS in 18m 42s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/247572b38f664b639f5273d2286e42ec : FAILURE in 32m 42s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4132027257bc4ef78718d094c6a5c68a : SUCCESS in 27m 40s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ed9b6c87413e4d3aa193c6d5ac488f66 : SUCCESS in 25m 39s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/82cf762d130d46aab34d2e0cd8f30b14 : SUCCESS in 14m 07s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d45a9b877dc4be989f22b889edc8a9f : SUCCESS in 11m 02s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/5ed0542875424e779b0b561bc58d561e : SUCCESS in 19m 08s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/96873afd2518463e8acae72626f2f9b2 : SUCCESS in 15m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5163300231564a1fa067db061134c8a3 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a2b358756144255b1e6308acc258a34 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26948a4c171540d1b82490bb95e76477 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":3},{"id":"9d0038bb1ef8fe575789f50b913945cde7c7434a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-03-07 13:42:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"308aaf3f1f3094beadb0f3f9ee92b8ea839a6d86","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-03-07 20:20:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0abefa0ebb9e3362060e438e655b5e906429999a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-01 09:06:06.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"e3d65b5afb5e3f3c2660628e48e3b893656a46f4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-01 09:15:26.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"34c76c4a82d480fbfff5704e4dcaac0a2068bbfd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-31 09:21:11.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"89b8d9d03618468f91e83943acd8a3ec87f42b23","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-31 09:32:57.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e569df05f8fef08e01e1ec4bcd8b9f61c3487d0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-31 10:22:39.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3b3129f42a240a5f4a6a47f3fd4de84b35adcdc9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-31 10:33:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8321816429c4d444e9b24fa4028a9d1f429a2c20","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-31 10:56:53.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"2f39ec083219938fc855c29343a441d9be74b458","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-31 11:07:42.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":5},{"id":"286c498b00dba3d278aaf79b6e7a5d00d156e153","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-31 11:08:38.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"468e73bbef1dd1cb0a1534aedaa6825809339ecd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-31 11:14:12.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e0e14f2d31fb4e7e01ab3ba9f1b3c45f307c317a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-31 13:00:27.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"a8b90a304ec6cfce0db87169ec3998ed5ab57730","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-31 13:15:02.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2806626f53d18884499fc8d9fb9874280d65e71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 15:09:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170084d781094dd992016b9c02bcb79a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d63da74e29648eb8436e0346d7462d0 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b12f05b3434f3cb10782b6c46c0f26 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3600a9fa6a83444da5a3b0b2a10406b6 : SUCCESS in 4m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b6625fe8f9174acea56d26f300c1186b : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9765e66fd69f442aa52e1426fe52f161 : SUCCESS in 10m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7399827e9d64035b095cf458a748837 : SUCCESS in 9m 19s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/068fceb0e0014074b31a914285bb812a : SUCCESS in 1h 00m 13s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/c8fc8b3cedde47e28d04fbeedbbd6e58 : SUCCESS in 1h 06m 22s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/46d06dd9f1204538aba6e42469b531d3 : SUCCESS in 49m 47s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b36809902db9475b90f3370d9e17cbc9 : SUCCESS in 1h 14m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5289d5f92ac74ea4812717dc591bca65 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/532c6341336140ccacdccd30b6dd7dbf : SUCCESS in 14m 33s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8a1a0da1e2e43588402df814c1ece00 : SUCCESS in 14m 44s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5d5016fce3a946d2970078e640e41d95 : SUCCESS in 2h 06m 42s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b284702f8c8c436fb6a1be03f9a3f5d1 : SUCCESS in 2h 03m 11s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/6db30c40e2fb448b9078f59b8a4b20be : FAILURE in 1h 25m 27s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c366236983e24f1aa9c27abc4d77ba4c : FAILURE in 59m 18s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/741396ec7b494e42ae241e4cef13f7cf : SUCCESS in 15m 47s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f35e64a5bdf242b6b07cb973453a3958 : SUCCESS in 14m 46s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/97d2b0ef5f714d0e9e6ae0044bc9e5e8 : SUCCESS in 35m 43s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/9c77511daa124f63b935d52ca8dc3a6e : SUCCESS in 31m 20s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8d073a8b920a4c00b1bc3af000809f5a : SUCCESS in 19m 30s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/c0bd831c296a492e8c11931bdd7901a0 : SUCCESS in 30m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1458387d3b60415b862987d29f5515a7 : SUCCESS in 25m 54s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/376a68df8c7740b686c17f90bc5ab068 : SUCCESS in 13m 17s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/00ca73a9309a4b4e97ed3a16e2f4555a : SUCCESS in 10m 19s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/14e515de361b4386a4fb442320316022 : SUCCESS in 17m 47s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/80e619eb00584971b866136b96196065 : SUCCESS in 15m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4705a24304274072b34ae7a47ad06c2d : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55a14a5abf5e42b7ac40c2d87d06cf41 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0f6986477ac446d88d384577552d9fd8 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":6},{"id":"138bf3918188697d6e95956eee1c678790b305b2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-03 08:57:44.000000000","message":"Patch Set 6: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8592a5b8e5c3e03d6200d720d552bc1e0cc1262b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-05 16:00:31.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"ed4d042d5e7922667008350fb97be405bacbaee5","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-05 16:02:26.000000000","message":"Patch Set 7: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"7e4354023cd1b50015054e7fae169aeafd707afd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-06-05 16:03:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f8e442897035bdf3e7e62755478f31e2ecd2e76a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 18:04:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82c2d6fe1e92475798e0d90d600e7a60\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1466bab46f594eae841d95b5069ab046 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8f8a991b1d04ace9c432479f10542d4 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99a7eed45ba5419a97d257633d8e7d70 : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/98952cd732374a609efee25bd8842a33 : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c5c7ea5275df4ba5b27406d1f29ff12a : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8911412c5cc4eaba331992c46358951 : SUCCESS in 9m 54s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/366291325f1640839cc69c4c93b5cfc9 : SUCCESS in 1h 04m 54s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1fd6c3362717466c93088c5324c693f5 : SUCCESS in 1h 05m 33s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1400be0d8e784280b33916564b07f634 : SUCCESS in 53m 20s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/b38c9e332e8b4b6eade2bc2b104aa55b : SUCCESS in 1h 09m 05s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/e097e02d45494cabb01326760dc24129 : SUCCESS in 17m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/a432da274bd240d09fa340de5baeeb61 : SUCCESS in 18m 50s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5762c149507a43628163083e57546848 : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/5161ff06afe042d68725689a9b0f3504 : SUCCESS in 1h 58m 14s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bf742492b6e4867a385a1edefac55fc : SUCCESS in 1h 38m 29s\n- kayobe-overcloud-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/13756036bbe446b2a568e2f4f8d45fc2 : FAILURE in 1h 29m 08s (non-voting)\n- kayobe-overcloud-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69f51c884777499aa0a55c9d4187c141 : FAILURE in 1h 11m 45s (non-voting)\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e4c3c5d0c12d463b833f6a1ab53e3587 : SUCCESS in 16m 59s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c70963896f14db18d3315234d1c438b : SUCCESS in 13m 51s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/30d3302cfe6c4d148fe61cf54e33828d : SUCCESS in 39m 26s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/533a07638ca244d2ba53008e9255d533 : SUCCESS in 27m 30s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fe992ce5e73f4f7198d2ec90cc8a8bd9 : SUCCESS in 26m 49s\n- kayobe-seed-upgrade-slurp-rocky9 https://zuul.opendev.org/t/openstack/build/1ae9f94730124eedad55b94357e912b8 : SUCCESS in 31m 58s\n- kayobe-seed-upgrade-slurp-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/898003c9a4bb49d3b69e8b93b0dce58d : SUCCESS in 27m 43s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/cea64dd035024a23b623de9e308599e4 : SUCCESS in 13m 25s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3ca250aaf1c44cce8d8f79a25d050e5f : SUCCESS in 10m 40s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/446c57cb1b1149e3bc0756f19ae1a63c : SUCCESS in 17m 49s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/527834b5f940485bb31554fdf6ffd9b5 : SUCCESS in 15m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4591971fb6034e529a5a4ed2ab787429 : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3de5a7d60f184e409635837d379ea9ae : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef0641a692724a42bb04189bede94d9a : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":7},{"id":"c5d2dc8df2034c2b0d6cbfabe79284775df12360","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-10-16 12:00:07.000000000","message":"Patch Set 7: Review-Priority-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"a80401f3d666bc06be03aa2e2705413378886ee8","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-10-16 12:00:13.000000000","message":"Patch Set 7: -Review-Priority Code-Review-1","accounts_in_message":[],"_revision_number":7},{"id":"8ad0b8ede2027926f3b5805a3f0eb43531bd83b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-25 16:23:38.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"f05813247c2b8df963bcb55320fd467734403487","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-25 16:28:54.000000000","message":"Patch Set 8: Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"7eb74d270423c9c2c1013bd03894b9b1e240ac1e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-25 16:31:01.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ecbd379c3d303d1c10210d940170be0bf00bcfd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-25 18:31:57.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f7c99d47ef245b5afa67833eda98ad3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2811dd4ac34478facc48ca5f5a3780b : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c4b7877c614b08be0fcbf6d540beba : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6dc96a19f294687be4e7fd8df859ec3 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85274c34cd0b4ac68664f4d69dfb315c : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99e40c8f3bf14a29b081c58bb1850503 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f72cfa3662f4dd9bdc9563e24bc4c38 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e5028a92e844498ca57fde3e1466dd01 : SUCCESS in 3m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a7f248298af48aea07e0c4e43ec930b : SUCCESS in 9m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4fbce702d4594f33af476dde17b04eb5 : SUCCESS in 6m 10s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ca6452077de145c7acf64006b0502518 : SUCCESS in 54m 27s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/bd3a8f40af2145698023491345279fa6 : SUCCESS in 1h 09m 16s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a4632d7e955b4904909b12ddfb4bba7c : SUCCESS in 51m 38s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/ff09e10eb54347c6a6d57ec627423916 : SUCCESS in 1h 13m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1493461953b1447199e3242fa4e9f5b0 : SUCCESS in 20m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/9023a36b94e24236963fb632239d2341 : SUCCESS in 17m 19s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e90d0f3826f3443c8f6eaf4426723b46 : FAILURE in 8m 50s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/4054719f7fa640f09f03916a4e3d2938 : SUCCESS in 2h 02m 22s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e86e10bb1106450ea6a61301fcd97798 : SUCCESS in 1h 32m 50s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/402f50214a2e4e7c9c4116446d6e88fb : SUCCESS in 18m 48s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5df04cf04864a5495f4e2cbe29188a0 : SUCCESS in 19m 32s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/bbeb1043cd944f6ba6c7e0d1338e8d23 : SUCCESS in 35m 52s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/2c57b1b3a16a4e49bb8dc85021c1e2a2 : SUCCESS in 38m 18s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e1a6615e7cf42ebb38034f605cf0dd4 : SUCCESS in 21m 59s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4e0d59e6905245e28b90bbae707d4d12 : SUCCESS in 14m 29s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e37de05e8e66431c9113d84c1b1b7a14 : SUCCESS in 16m 53s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f658be05eebb434aa5749028f6d148fb : SUCCESS in 14m 18s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/c8542db3abf94aa8bf4989927d432d8f : SUCCESS in 10m 57s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/414e2ecd23c84783a85eef099ff1aed7 : SUCCESS in 15m 18s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/99e09b67f1e44ab292739b0cbe8649eb : SUCCESS in 16m 03s","accounts_in_message":[],"_revision_number":8},{"id":"6d26262582ba2c9476595ce9b31cee00a95354d1","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-10-30 18:15:23.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ecf92a693bcf65c1c959a32e61570a59b2eaa508","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-10-31 14:52:48.000000000","message":"Patch Set 8: -Review-Priority\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"784c7bfce260dea5207a1a9b1b23b70d91f558ad","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-10-31 14:56:59.000000000","message":"Patch Set 8: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"39ff45b7e3f401afce104d3e52c3405fbc507d55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 17:09:48.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a9c70f389154dcc8984eea3791f8244\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd3d1791ffd0444da449b029bfe098b4 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdca6bc3506a4d2e90871bf02f57f5b3 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03f3d841da574d309ddfb72548f9b22f : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ee7372519e34f5ca8a64d4eed7b88a3 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8441d32a7b744f5b859fe6fe17abf0f2 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1567e01670404154a01f8c7c5536ab00 : SUCCESS in 2m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/34b10ab621544f07905cda0b467dbd71 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f70ce99a554c4a439c6fc7836bfd25f1 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b8eca235126452ea20ad07e5d3cce52 : SUCCESS in 8m 45s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/57e72cd2d144484281c8c43c5c8a42b9 : SUCCESS in 58m 52s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/202c0eee99b54f949f7ddfbcfc8053d0 : SUCCESS in 57m 31s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f73ea23c945744c2a98b486584b14ef1 : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/897cdec7bcb64854aa80945c638c545e : SUCCESS in 1h 02m 43s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c7f71a4a85084f15b67640e1f790c417 : SUCCESS in 20m 43s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/c4cfe03dd18f47578759b3aacc7642c9 : SUCCESS in 11m 37s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-noble https://zuul.opendev.org/t/openstack/build/482f4456433743a8aa63bbfa584b2a1f : SUCCESS in 10m 56s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/bdf8c0059ea2405d95afdcf42a13da4d : SUCCESS in 1h 50m 19s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ab6011d636ca4ff8a378f16aab35cda7 : SUCCESS in 1h 36m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/3ba13341674348c5959057d8a880be86 : SUCCESS in 19m 39s\n- kayobe-seed-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7542c0d5938c4625a7725b093405ca72 : SUCCESS in 9m 30s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/ccbfdc9707e747ccaf1fe76184ccf7c5 : SUCCESS in 27m 35s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/345bf4971baf410391fd726ee797d7ad : SUCCESS in 36m 26s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d69df2df3d4143b885c026b428ce505c : SUCCESS in 26m 19s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/ab920fe9499e421090dc215ea0156d05 : SUCCESS in 18m 48s\n- kayobe-seed-vm-rocky9-efi https://zuul.opendev.org/t/openstack/build/e59d5b638c134d4f92a4b5cc8e9bf257 : SUCCESS in 16m 41s\n- kayobe-seed-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6deabfd948284e2e97aea04c44bb9ce6 : SUCCESS in 7m 41s\n- kayobe-seed-vm-ubuntu-noble-efi https://zuul.opendev.org/t/openstack/build/28d5f1c339c44023a6ea397dc2d7de6a : SUCCESS in 10m 37s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/04cbd14c63e9403c885c11b113d7499a : SUCCESS in 16m 15s\n- kayobe-infra-vm-ubuntu-noble https://zuul.opendev.org/t/openstack/build/69d1a0bcf83244168e68e0b07f307565 : SUCCESS in 15m 09s","accounts_in_message":[],"_revision_number":8},{"id":"8ed203de018f753d2089f3fbf6235cd3bb19370f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-31 22:39:12.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"ada9fe7a5347894889dd2ead75397d0f1b63ce81","revisions":{"096899c3cb7643fd54433a97bf4d893d45bb32e2":{"kind":"REWORK","_number":1,"created":"2020-01-21 18:45:25.000000000","uploader":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"ref":"refs/changes/67/703667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/703667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/703667/1"}}},"commit":{"parents":[{"commit":"b5749675a3fa22ce80a8c347a64928b176fa233d","subject":"Add proxy support to Vagrantfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5749675a3fa22ce80a8c347a64928b176fa233d"}]}],"author":{"name":"Victor Morales","email":"victor.morales@intel.com","date":"2020-01-21 18:39:27.000000000","tz":-480},"committer":{"name":"Victor Morales","email":"victor.morales@intel.com","date":"2020-01-21 18:39:27.000000000","tz":-480},"subject":"Add vagrant-libvirt support","message":"Add vagrant-libvirt support\n\nThe Vagrant Libvirt provider[1] offers support for the creation of\nVirtual Machines on Libvirt. This change adds the instructions\nrequired by that provider\n\n[1] https://github.com/vagrant-libvirt/vagrant-libvirt\n\nChange-Id: I7c65689808cc0bf592a06b37f570db5b5cf7480c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/096899c3cb7643fd54433a97bf4d893d45bb32e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/096899c3cb7643fd54433a97bf4d893d45bb32e2"}]},"branch":"refs/heads/master"},"7a8ab15b3449b198f0c2cbbf417c0a27599b6252":{"kind":"REWORK","_number":2,"created":"2024-02-20 15:07:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/703667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/703667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/703667/2"}}},"commit":{"parents":[{"commit":"f2b658ecb6c0ba439f6e760bb8aed3e1d6779ed5","subject":"Merge \"Generate KA config before ``kolla ansible run``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2b658ecb6c0ba439f6e760bb8aed3e1d6779ed5"}]}],"author":{"name":"Victor Morales","email":"victor.morales@intel.com","date":"2020-01-21 18:39:27.000000000","tz":-480},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-20 15:07:44.000000000","tz":0},"subject":"Add vagrant-libvirt support","message":"Add vagrant-libvirt support\n\nThe Vagrant Libvirt provider[1] offers support for the creation of\nVirtual Machines on Libvirt. This change adds the instructions\nrequired by that provider\n\n[1] https://github.com/vagrant-libvirt/vagrant-libvirt\n\nChange-Id: I7c65689808cc0bf592a06b37f570db5b5cf7480c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a8ab15b3449b198f0c2cbbf417c0a27599b6252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a8ab15b3449b198f0c2cbbf417c0a27599b6252"}]},"branch":"refs/heads/master"},"1d42588827cf86978d29930c877304dbfdb2a677":{"kind":"REWORK","_number":3,"created":"2024-02-20 16:59:16.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/703667/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/703667/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/703667/3"}}},"commit":{"parents":[{"commit":"f2b658ecb6c0ba439f6e760bb8aed3e1d6779ed5","subject":"Merge \"Generate KA config before ``kolla ansible run``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2b658ecb6c0ba439f6e760bb8aed3e1d6779ed5"}]}],"author":{"name":"Victor Morales","email":"victor.morales@intel.com","date":"2020-01-21 18:39:27.000000000","tz":-480},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-02-20 16:59:15.000000000","tz":0},"subject":"Add vagrant-libvirt support","message":"Add vagrant-libvirt support\n\nThe Vagrant Libvirt provider[1] offers support for the creation of\nVirtual Machines on Libvirt. This change adds the instructions\nrequired by that provider\n\n[1] https://github.com/vagrant-libvirt/vagrant-libvirt\n\nChange-Id: I7c65689808cc0bf592a06b37f570db5b5cf7480c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d42588827cf86978d29930c877304dbfdb2a677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d42588827cf86978d29930c877304dbfdb2a677"}]},"branch":"refs/heads/master"},"020d87d0f3e7c9be45cbf9e3dd9243f75ae7f37b":{"kind":"REWORK","_number":4,"created":"2024-05-31 09:21:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/67/703667/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/703667/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/703667/4"}}},"commit":{"parents":[{"commit":"f2b658ecb6c0ba439f6e760bb8aed3e1d6779ed5","subject":"Merge \"Generate KA config before ``kolla ansible run``\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2b658ecb6c0ba439f6e760bb8aed3e1d6779ed5"}]}],"author":{"name":"Victor Morales","email":"victor.morales@intel.com","date":"2020-01-21 18:39:27.000000000","tz":-480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-05-31 09:20:50.000000000","tz":120},"subject":"Add vagrant-libvirt support","message":"Add vagrant-libvirt support\n\nThe Vagrant Libvirt provider[1] offers support for the creation of\nVirtual Machines on Libvirt. This change adds the instructions\nrequired by that provider\n\n[1] https://github.com/vagrant-libvirt/vagrant-libvirt\n\nChange-Id: I7c65689808cc0bf592a06b37f570db5b5cf7480c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/020d87d0f3e7c9be45cbf9e3dd9243f75ae7f37b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/020d87d0f3e7c9be45cbf9e3dd9243f75ae7f37b"}]},"branch":"refs/heads/master"},"924d3e187811ac215ca3fb11b136b65380f14730":{"kind":"REWORK","_number":5,"created":"2024-05-31 11:07:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/67/703667/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/703667/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/703667/5"}}},"commit":{"parents":[{"commit":"d17f7dc1b8acb447a556d6cdaea69cf46c0daf27","subject":"Merge \"docs: Reword bare metal registration section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d17f7dc1b8acb447a556d6cdaea69cf46c0daf27"}]}],"author":{"name":"Victor Morales","email":"victor.morales@intel.com","date":"2020-01-21 18:39:27.000000000","tz":-480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-05-31 11:02:20.000000000","tz":120},"subject":"Add vagrant-libvirt support","message":"Add vagrant-libvirt support\n\nThe Vagrant Libvirt provider[1] offers support for the creation of\nVirtual Machines on Libvirt. This change adds the instructions\nrequired by that provider\n\n[1] https://github.com/vagrant-libvirt/vagrant-libvirt\n\nChange-Id: I7c65689808cc0bf592a06b37f570db5b5cf7480c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/924d3e187811ac215ca3fb11b136b65380f14730"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/924d3e187811ac215ca3fb11b136b65380f14730"}]},"branch":"refs/heads/master"},"fa568822b86add8e198bea965f2a3a686119f84f":{"kind":"REWORK","_number":6,"created":"2024-05-31 13:00:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/67/703667/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/703667/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/703667/6"}}},"commit":{"parents":[{"commit":"d17f7dc1b8acb447a556d6cdaea69cf46c0daf27","subject":"Merge \"docs: Reword bare metal registration section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d17f7dc1b8acb447a556d6cdaea69cf46c0daf27"}]}],"author":{"name":"Victor Morales","email":"victor.morales@intel.com","date":"2020-01-21 18:39:27.000000000","tz":-480},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-05-31 12:59:13.000000000","tz":120},"subject":"Add vagrant-libvirt support","message":"Add vagrant-libvirt support\n\nThe Vagrant Libvirt provider[1] offers support for the creation of\nvirtual machines on Libvirt. This change adds the instructions required\nby that provider.\n\nIt also updates the Vagrant box to Rocky Linux 9 and uses\nkayobe-config-dev as configuration repository.\n\n[1] https://github.com/vagrant-libvirt/vagrant-libvirt\n\nChange-Id: I7c65689808cc0bf592a06b37f570db5b5cf7480c\nCo-Authored-By: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa568822b86add8e198bea965f2a3a686119f84f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fa568822b86add8e198bea965f2a3a686119f84f"}]},"branch":"refs/heads/master"},"c21e5647b90f1615cb3f0351541ebe4d313ff503":{"kind":"REWORK","_number":7,"created":"2024-06-05 16:00:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/703667/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/703667/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/703667/7"}}},"commit":{"parents":[{"commit":"d17f7dc1b8acb447a556d6cdaea69cf46c0daf27","subject":"Merge \"docs: Reword bare metal registration section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d17f7dc1b8acb447a556d6cdaea69cf46c0daf27"}]}],"author":{"name":"Victor Morales","email":"victor.morales@intel.com","date":"2020-01-21 18:39:27.000000000","tz":-480},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-06-05 16:00:29.000000000","tz":60},"subject":"Add vagrant-libvirt support","message":"Add vagrant-libvirt support\n\nThe Vagrant Libvirt provider[1] offers support for the creation of\nvirtual machines on Libvirt. This change adds the instructions required\nby that provider.\n\nIt also updates the Vagrant box to Rocky Linux 9 and uses\nkayobe-config-dev as configuration repository.\n\n[1] https://github.com/vagrant-libvirt/vagrant-libvirt\n\nChange-Id: I7c65689808cc0bf592a06b37f570db5b5cf7480c\nCo-Authored-By: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c21e5647b90f1615cb3f0351541ebe4d313ff503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c21e5647b90f1615cb3f0351541ebe4d313ff503"}]},"branch":"refs/heads/master"},"ada9fe7a5347894889dd2ead75397d0f1b63ce81":{"kind":"REWORK","_number":8,"created":"2024-10-25 16:23:38.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/67/703667/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/703667/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/703667/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/703667/8"}}},"commit":{"parents":[{"commit":"d17f7dc1b8acb447a556d6cdaea69cf46c0daf27","subject":"Merge \"docs: Reword bare metal registration section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d17f7dc1b8acb447a556d6cdaea69cf46c0daf27"}]}],"author":{"name":"Victor Morales","email":"victor.morales@intel.com","date":"2020-01-21 18:39:27.000000000","tz":-480},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-10-25 16:23:37.000000000","tz":0},"subject":"Add vagrant-libvirt support","message":"Add vagrant-libvirt support\n\nThe Vagrant Libvirt provider[1] offers support for the creation of\nvirtual machines on Libvirt. This change adds the instructions required\nby that provider.\n\nIt also updates the Vagrant box to Rocky Linux 9 and uses\nkayobe-config-dev as configuration repository.\n\n[1] https://github.com/vagrant-libvirt/vagrant-libvirt\n\nChange-Id: I7c65689808cc0bf592a06b37f570db5b5cf7480c\nCo-Authored-By: Pierre Riteau \u003cpierre@stackhpc.com\u003e\nCo-Authored-By: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ada9fe7a5347894889dd2ead75397d0f1b63ce81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ada9fe7a5347894889dd2ead75397d0f1b63ce81"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
