)]}'
{"id":"openstack%2Fkayobe~704611","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I5dccd7d8eab8d46ef5e8cd97a601510a3d8db177","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"I5dccd7d8eab8d46ef5e8cd97a601510a3d8db177","subject":"Don\u0027t list tasks for config dump","status":"MERGED","created":"2020-01-28 16:16:54.000000000","updated":"2020-02-10 18:28:02.000000000","submitted":"2020-02-10 18:28:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704611-1581359282614-25e6796c","meta_rev_id":"b95c208d77dcf964ecf04d924005789abe23e731","_number":704611,"virtual_id_number":704611,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 18:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-02-10 12:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-10 12:08:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-02-10 12:11:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-02-07 21:52:32.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-07 17:38:48.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-02-10 12:08:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-02-10 18:28:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"743a3f03d4f3ef2d558bcecffeee4b03fcba5204","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-01-28 16:16:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0eaf77208de4eff4fc7d5e6d5209b4ede3ebd70a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 18:03:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/60453b3dbdca4710b9615b817594f21e : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3066596467344d3b574383e5bd0047f : SUCCESS in 4m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec08331b589c40d5b2df11542b48b53b : SUCCESS in 6m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1291030f06384c43b069778bf58d33a0 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e14fdef8ec074d3f80184673f4172df1 : SUCCESS in 5m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8868df8756fd4b62bcbb60da3e6ebbaa : SUCCESS in 5m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/44bfbff7ca0243d49eb6a01d0d2743fc : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/99492a1d99584495978a1e7065d798a7 : FAILURE in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d22c0ca3c8d54a1bb171a098990d7328 : SUCCESS in 36m 08s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0184d534458e49d993867296311ba217 : FAILURE in 11m 34s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/de5a84c5843740588a564f6893ae87d6 : SUCCESS in 1h 34m 55s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/780632e158574141abeed8cad506b5f5 : FAILURE in 10m 21s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c62f88d315484e43b15e51d4a67b3f22 : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c38a47d109c8bbe8c8e65fedfa5a66891b47f954","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-07 21:52:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"911916ae2aac54b54625a9534b59be70f0691240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 23:41:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/9fd8f075723f4450b34ad1ab3678340d : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e742a0e084443ca72fd35ac2de89cb : SUCCESS in 4m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b72796e6280141fca92adc20c105fba3 : SUCCESS in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a10ffd93b00146f084f4f343448215f7 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0362c0314b7b4323bd4fddfa5003bf99 : SUCCESS in 5m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e3cbb855e9c240c6bf51e89ecbc963ab : SUCCESS in 5m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e0b124297143498fabe85c11a995d958 : SUCCESS in 8m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/134ee730c2a844eea4de462590f7750b : SUCCESS in 13m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e4a746a73154526af86ebd7bbde5adc : SUCCESS in 38m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b6b148f3bfc7488589296221cdde56bc : SUCCESS in 58m 33s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/83888e2529de47b981651a7679a08d80 : SUCCESS in 1h 43m 59s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/d01fe2f5666b4da293a803f81d06096e : SUCCESS in 23m 40s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/070cb990704248bebad860602bc88117 : SUCCESS in 44m 12s","accounts_in_message":[],"_revision_number":2},{"id":"3d09aa620601ea6bee8dcd2e797d0a36a6a013ab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-10 12:08:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"67a7f072b4544e61c53dcdc55a43c06b1d1f5e0f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 12:11:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0549b2a6b9d296d7e60f6f1f910faa31fbfbd687","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 12:11:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"18e7dcce0fbc7ecfdbbf9b5765df7285c1087b0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:13:20.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/189e8976be5b4684808e7a5036fdf6a4 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3314330175f34352961727245192606a : SUCCESS in 6m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/83e060a143a6431aa0bc0516d17628f6 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81fd939461544c40a0d21ca45668dc6f : SUCCESS in 6m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f862e89a6bb34449923c4f3e3e6e1c74 : SUCCESS in 5m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2120ddb9bd9f4b269df51520c2c9b03d : SUCCESS in 7m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f66fcf27b3934cd281a3c46ecb6196ca : SUCCESS in 9m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/16968409480248f883cd8fd5247aaf63 : SUCCESS in 29m 05s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7eceef09f6074640b8f4e0e74da81b8a : SUCCESS in 1h 00m 14s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/75c0b5d7648b43709562b6e0d5e0c9d1 : FAILURE in 44m 41s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/6679c06c0ee44b26a42cbbe431a06b25 : SUCCESS in 21m 26s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/86f6cb77d93b4c52930a5889e23c93c3 : SUCCESS in 42m 00s","accounts_in_message":[],"_revision_number":2},{"id":"f8a327e3b118de239357c3249269f020af8d05f0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-10 14:33:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"0b79b6cac0854bdfcb4e0244f3cb67d1a792909f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 16:25:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/ca59152d304d48638fd2e6a46774bc6b : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0fb484f3e9149fe97ce2c46475eae6b : SUCCESS in 4m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67c9f0b1184447019d33056c3079d946 : SUCCESS in 8m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c7b49d238172420da75382d78ed21e58 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa968468432e4d9fa07d8b6ecd1596aa : SUCCESS in 7m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c7ab7876e190460f88517f4712e3735e : SUCCESS in 7m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c15e38055364117a9c46aa9d8c6ccd6 : SUCCESS in 7m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35499f407fc742d4a0f9765413414c3e : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eeac28b452af434bbc94b6fcebaa347a : SUCCESS in 34m 42s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/a0fd50e3030a467faaf2c4e237e0e6b1 : SUCCESS in 1h 02m 09s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/24846acd30504d28af3609e362cd78a4 : SUCCESS in 1h 45m 17s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1ab6c3f0cec04bdba2a2786fb0913384 : SUCCESS in 21m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/563e5d3722604a07b41ad9160a68a122 : SUCCESS in 41m 09s","accounts_in_message":[],"_revision_number":2},{"id":"7efb9cce39a8ced7b3a5f48bb9cbe6166e027a7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 16:26:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b846cbb0d2198a395a2a3d53f3a4f19f9709b767","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:28:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b95c208d77dcf964ecf04d924005789abe23e731","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 18:28:02.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/4c41f5f8399f4b0cb5ffe3c271c15aa6 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a922a3867084b4e9dbda42a38c8e11a : SUCCESS in 4m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cbbc712afac848d5b6652bc7f704d0e6 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/898cc1531db741278a8a268f5e8384ee : SUCCESS in 6m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/010f12aeccbd479d9223fc53d1151de0 : SUCCESS in 7m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/96a5891246ed4ac2a2c057e9881ddcbd : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5be9c1a861e446d486df3964559298d8 : SUCCESS in 8m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/caf3419ce8cb4c86a528480f26f20edc : SUCCESS in 34m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/55fe2ebdfa104959880fb19f10d41548 : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/58725d5411ef491abe1631e6582d3eff : SUCCESS in 1h 56m 44s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3dbe05f9ebab4e8190b4920a6a8939ac : SUCCESS in 24m 35s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/54df9c00a6d343118ae0524277dfe295 : SUCCESS in 46m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1288908b4c6840041a781c1797d070a082b71c06","revisions":{"d55db597aca551c9a000b45ccf98a1c10c55867a":{"kind":"REWORK","_number":1,"created":"2020-01-28 16:16:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/11/704611/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/704611/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/704611/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/704611/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/704611/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/704611/1"}}},"commit":{"parents":[{"commit":"4652d3aae3c9c7790542283036292622e91b7cf3","subject":"Add support for custom Elasticsearch config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4652d3aae3c9c7790542283036292622e91b7cf3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-11 17:54:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-01-28 16:16:54.000000000","tz":0},"subject":"Don\u0027t list tasks for config dump","message":"Don\u0027t list tasks for config dump\n\nkayobe executed with --list-tasks often fails due to passing this\nargument through to the config dump, which causes it to not generate the\nrequired dump file.\n\nIgnore list-tasks for configuration dump.\n\nChange-Id: I5dccd7d8eab8d46ef5e8cd97a601510a3d8db177\nStory: 2006530\nTask: 36584\n(cherry picked from commit 844202d226c9d2ed64c18e7d1b36de131021b4b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d55db597aca551c9a000b45ccf98a1c10c55867a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d55db597aca551c9a000b45ccf98a1c10c55867a"}]},"branch":"refs/heads/stable/rocky"},"1288908b4c6840041a781c1797d070a082b71c06":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-07 21:52:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/704611/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/704611/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/704611/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/704611/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/704611/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/704611/2"}}},"commit":{"parents":[{"commit":"b57ff069d27b9fac4b8bcb9a7454205fca1c63aa","subject":"Fix multiple CI failures","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b57ff069d27b9fac4b8bcb9a7454205fca1c63aa"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-09-11 17:54:12.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-07 21:52:32.000000000","tz":0},"subject":"Don\u0027t list tasks for config dump","message":"Don\u0027t list tasks for config dump\n\nkayobe executed with --list-tasks often fails due to passing this\nargument through to the config dump, which causes it to not generate the\nrequired dump file.\n\nIgnore list-tasks for configuration dump.\n\nChange-Id: I5dccd7d8eab8d46ef5e8cd97a601510a3d8db177\nStory: 2006530\nTask: 36584\n(cherry picked from commit 844202d226c9d2ed64c18e7d1b36de131021b4b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1288908b4c6840041a781c1797d070a082b71c06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1288908b4c6840041a781c1797d070a082b71c06"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
